)]}'
{"id":"openstack%2Fswift~940791","triplet_id":"openstack%2Fswift~master~Iaf86a07238cca6700dee736f55d4c0672cccf1b1","project":"openstack/swift","branch":"master","topic":"s3helper","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-11-03 04:31:19.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-10-13 15:33:59.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-11-03 05:37:41.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-11-03 05:37:41.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-11-03 05:37:41.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-03 05:37:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf86a07238cca6700dee736f55d4c0672cccf1b1","subject":"Provide some s3 helper methods for other middlewares to use.","status":"MERGED","created":"2025-02-05 15:00:34.000000000","updated":"2025-11-03 05:38:33.000000000","submitted":"2025-11-03 05:37:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":203,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940791-s3helper","meta_rev_id":"e7624020c9d106844b69bcc91358465bea99470f","_number":940791,"virtual_id_number":940791,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-03 05:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-11-03 04:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-10-14 13:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-11-03 04:31:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 16:21:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-07 14:57:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-05-08 19:56:32.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2025-07-16 20:46:50.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2025-08-07 21:49:03.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-08-25 05:27:56.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-09-24 17:32:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"82a425417ac2473beaf310642216eca49ed63e2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-05 15:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf82fa8f18bb7f5af0ffe5ee901a162ded84ff68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 16:21:51.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d93c65177b54874985a869e857684fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263cdf04adab4202b861904aad710b2e : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/712b052e80e34715afcc8bae1df71b45 : SUCCESS in 57m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/384d3ab5d4594ef8aa08b372a6bd08bd : SUCCESS in 1h 14m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/030835ac4f9b4b9dabb3268f35ec4454 : SUCCESS in 56m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eda9d9e7da84bceb6864d6e72deb057 : SUCCESS in 1h 06m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00fc6f4921864ce1b760a2a51b871de8 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de4742a9226940a081d43767b598ef1b : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11033dab59e04f5f8aa99a89d50e1335 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d99313c5d0ce4ea6ae4dc510f0a80bbf : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/008bd4ec5fc64b2f80b2d4cccb613a10 : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c816aa95f2704b44bb85d0ba90f231b5 : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01c12e55681647dd8802202a24a7519e : SUCCESS in 26m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/193311f31a6a4bce85171544bea0bfd3 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4b674be612424105940af7fa96afae31 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/139c0b3490df4bd183d17390d1ee6715 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f84ef4d144548bc9c2bb055f91bd3d1 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/986b9e1ab86b4ceabc045d7a87c078ba : FAILURE in 14m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfcf2a3a4f38466680883d45e586b4ff : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c7a3f32fb664eb0946290af42b604c9 : SUCCESS in 53m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6a54ad0bffe435e920fdcef114df684 : SUCCESS in 45m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98fa20dd6d2a443f800f592a95d661c6 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bdc0bddedf04563a7dcf50d64cc5590 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5234d6177c1c48a39ffb8025a2e7ea78 : FAILURE in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4ea40d5ef8d41719c504c3e417644d4 : SUCCESS in 17m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10d491db238247b3a9c47aa73c7944b7 : SUCCESS in 52m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a1641ba2455465e83002d9966c0434d : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4cb1a489fd815cb40f0609f202d7fb3a62619921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 16:25:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcbba85c81e54c728d6a7eec657eaffa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8902f17f416c473eb29f817ff167d6f4 : SUCCESS in 1h 22m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4e28835da6d07f48a02e99b66e999524311e228d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-07 14:57:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d73c0c38d2981eb9549df34613735003491340af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-10 15:03:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"133e521b0db557e8b16096c5af0dea919539a7ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 16:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b49a41308224dac9e484dcaaf91741d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a314dfa426e64e039d32b632d9ad072e : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/7ec92d99b73346e3b89c3ff348b6f485 : SUCCESS in 54m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6103e783bd2d4090b0ccabcf5ab3b939 : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b4aad5acb1824f2087f76953c1f16d01 : SUCCESS in 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66a0ae243de54b4a80eddae7ca4d8ad3 : SUCCESS in 54m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11bca29c6cea455bb71975f57c2af196 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/410554de062246008a451a456aadb38a : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9673f8ab346490795f278c2809af825 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f17119e5c0bc46368034777969f57bc3 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/264b5172880a4ff3becd31589320e149 : SUCCESS in 17m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8121cc9b130f4c8b98193805e0cdecb9 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/58e43d8be5f848169a13f387fbf8b7e2 : SUCCESS in 19m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb08dcf523824410a3d1f14c34c09d13 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8452ddd6c267469d9137cf770d299422 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40563dc465ad4a6f93c8e4f769e2fb4e : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c140b9718eb48519d512d7596240413 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0843676ff1d4abb8c6fadf9b05aa9ec : FAILURE in 13m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6959c5173e86410dbb7b20ddf7b87b75 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42e471436ef84fb98d0460c3e843be0a : SUCCESS in 50m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb756602564a4658ab72876e47562246 : SUCCESS in 1h 00m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40a23ebb51104831a585238997524827 : SUCCESS in 33m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a142cc0ba5d34d0e83c04ca33cd0fe78 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef5cda8b204450d8176783d86bf9f1b : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32033c5f2bb04edf86c7b65944149e7a : SUCCESS in 23m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107279dc47b547a996c2a72abefbc08c : SUCCESS in 1h 14m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/750f83cabdd94e1783c7ed21372e5a3e : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":2},{"id":"bf03b90b02cae1472c60aee39b213e452831e971","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 16:40:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05df701e4272490ab1cf366a284dc57c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc10a015ab8c49508e5b3cc3fac74a21 : SUCCESS in 1h 34m 57s","accounts_in_message":[],"_revision_number":2},{"id":"956f23bf26584a01a157ab59b5e5ee70e4b4ae10","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-12 12:06:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1452130ee18bce39f0fa3250b4b1e09de2f5ee18","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-12 16:21:39.000000000","message":"Patch Set 2:\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ddba8a3f0e96e32adf8a98256298f0e27d5e9024","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-13 20:31:58.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"974bc12fcfc22bcae2ae6a73cac87464c3f6ae18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-13 20:59:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"389d15e8e44c27a206bb84cbec3fcb4bcf6603a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 22:17:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a422782d59e440cbaa1db55faee10d2d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce7034c9597448238b0e787efbd2af72 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":3},{"id":"475e1fe70b02fba3ae05fdf501bb3ae250acde67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 22:22:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f633493bddb84560917a15f0fbd8a104\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb42f9a86ea4556a0f0c7186ccce4fc : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f497a9cbc463442183866c3af8d77b46 : SUCCESS in 49m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8df032395db49e984164cf782b13a04 : SUCCESS in 1h 02m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a435524e57745e98dfddfaf629d740b : SUCCESS in 44m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd8c4ccce93f4d7781b513cc76584c4b : SUCCESS in 52m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f798970a4a8468f8505c07704201701 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6be706a320aa47b19dda29c79ff93e0d : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c230073a3a4a40b693def0efbdb3ec97 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/243f2187bf22402c9fa87796826c16d5 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc5bd70729394255a8b2cb195cd3c6e5 : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f76d17f471243369d20a563a46e2f0b : SUCCESS in 26m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9ad1e079bc94ae2a4f0f4c0faca2f4c : SUCCESS in 14m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/277988da28864510a8a9bdc3d85662b2 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ad41108ffdcc46e8b2f81c371cac65a4 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b137a26406bb46089e21701d086b62ea : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01d0f36c80dc439684bb45daf2166424 : SUCCESS in 4m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63f6ab8b92f14c96b3cdf75bb2418814 : FAILURE in 18m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2975520eb4e94a189be7eb976a41ff1b : SUCCESS in 4m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a68777ee25fb4cbdb7724c906628e3a8 : SUCCESS in 37m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bddf246f18149119e0960da217212f6 : SUCCESS in 25m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c38b98cde7e149abb3e890c92b595b46 : SUCCESS in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07127e45bd39496e8c1f6667902614bf : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08601f4fafc4f12afddaa46c5620e2e : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c8c7e854c424edb9e9f9bb11cad89ef : SUCCESS in 22m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d4f5859bc249a1b28f9938fbd0ee77 : SUCCESS in 43m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7972b2de0a46cbb43ded78fef8e669 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":3},{"id":"1eeaa18031ee17dc70607e48f1663f1baf1e7e49","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-19 12:13:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":3},{"id":"73afd3b577c9c0a85fe2e0a7bc188da2b501debf","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-26 18:59:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"02059bab187801cbe6b8ecfee953991b0847386c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-27 15:00:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"423956fd54c631edcd84f3308e6b616a0907a442","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 17:41:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dace2abdb2cd40a2a7207cefbd436bb1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2f1e351f5a3d4788967bc2e3b909bc73 : SUCCESS in 2h 31m 51s","accounts_in_message":[],"_revision_number":4},{"id":"4f1760b20f030e1283fe7e31ed7c281c42fdb090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 00:31:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/762a372011454c9dbc1739894ee65187\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfa16b4b5b641cd87162132d68c52cf : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/1125072edb2947b9891f4b3328b24cd3 : SUCCESS in 47m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e886a16fc941440cb2bd45e8d90d9811 : SUCCESS in 45m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcd4ace1a8824503bde591955fbcad33 : SUCCESS in 46m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6a39b7273ff4ea4b32ce3b6db825176 : SUCCESS in 1h 55m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c27f5114322045f8982dd618d7d88fca : SUCCESS in 28m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a93078cad5a7473996f38db469500b8e : SUCCESS in 25m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c640a9e6515943c48d989caf25a193d7 : SUCCESS in 27m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2bc43179df04aabac1949b798f55d4d : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b6d0c1bc1fb4fd6bd4f7143df5a3763 : SUCCESS in 18m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9b2a6e0b3f643b6b56a73dbd31ce918 : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a375c56b68444c799e545c2443aca935 : SUCCESS in 26m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/adc4e739d95043d0aa0fb6e53a9d86ac : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e3a1611430b749d2a2215b5ce0c867b0 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1a896bcaec2e4309a1dadd47945b4888 : SUCCESS in 13m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38f81d9d9c62453c9a4c79cd0b25560e : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/efb27c17ce8c4e7b806bc2b6ec3a77af : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd59b7153ce343a6acc2b4d55b1c5f7a : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/32a342870b624ab78abb9c5ac051df3f : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/abe320178eff4816ad1878e57cb921b5 : RETRY_LIMIT in 12m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5502a000bf94f709d04ecdf4fda1cd6 : RETRY_LIMIT in 10m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9035207fd9f448b6b52233a43e1df9f3 : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789dff7172294cd5acf71ebb94222ff0 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/017db57d1a624fb88df82c03d2b0cc83 : SUCCESS in 19m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/605dc64f740c43d3bd4d5d40d2f95a31 : SUCCESS in 44m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04463d0efba74618b7105aa2af43b253 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":4},{"id":"c05a97c367526d4022a8d2d01c731e4c6a52a4d2","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-02-28 20:22:31.000000000","message":"Patch Set 4:\n\n(20 comments)","accounts_in_message":[],"_revision_number":4},{"id":"20bd020de83a4dd156f086b1df6f2400519235c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-05 14:58:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"52e52d7fe81709137f8b3f20e4af251a7fda1a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:09:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7cf209218f47e6a8596cf9c51fc67e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c901540ec7e4926a9d0cc43f7648370 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bdf29666bf3b4d8e91cb302ff178b480 : SUCCESS in 58m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/afbab9968568455d84c8325e3b865049 : SUCCESS in 56m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b898b95435e948ffb330be62249c2783 : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c367e99176147ee91c1177951d958dc : SUCCESS in 1h 04m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/668610e013d14b43bed512cf29bc7753 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d876d04d1474bfba96c3a158b7bf4c1 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9014777339c541b4ab2b52185f2020a7 : SUCCESS in 16m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb3849a03ce54f319ee1a49890f8c41b : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a8500d359a49d49711661470684235 : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef73e6d904df4ce5bbe2f09a89e32582 : SUCCESS in 20m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/78a70989bf284357bca57ea85260af40 : SUCCESS in 22m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1c91801c3e748979a4b529f743441f9 : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/34d6713a2f344629afb2749333066ae4 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc35cf10a2b1412c9430eca426bc4338 : SUCCESS in 14m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ab6dc4be85c4a18b1af193d2c39ea51 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e662b1992613492ea7e0c52380e06e59 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/861485b97d464ca3b427dd08cc44ea29 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22045af42eed4635ad147168144a3577 : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0af31a87630d4b9783ca4187ad9adaa2 : SUCCESS in 35m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93316a283aec43369a3c8ccfd75a271b : SUCCESS in 44m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af2de28bce0f4ecd86f43e24041ed5f2 : SUCCESS in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ffdce947d8478f92b84b6a511c50a4 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b1bb9492efb4b6dae8a9f0873c2fd5e : SUCCESS in 22m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cdb92d63f2646539ab545b69642a05e : SUCCESS in 49m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13c49fd45beb4706b55913af05f52ad5 : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":5},{"id":"489b170b81a90cfc3175c534fbbbd13813e6f22b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:44:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa5991e33ee47e8815f365678114c3e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fddd2de8728462389da8c9de9f162ad : SUCCESS in 1h 36m 55s","accounts_in_message":[],"_revision_number":5},{"id":"7ccb240973e0de42fe498e5582f44ae1dece1a32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 11:58:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7e5ac924a841e504a2067bade0d4ea428bd6356b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:07:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c15801747d4e558af6f4db421ee1f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a4197ba225f471787717ff0b7bc7104 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/ba2c0356eced41069c8f2c3e5533e31e : SUCCESS in 58m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db78cd85b65040e581472690d5199f03 : SUCCESS in 33m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b7f4ee0d639b4b38868615514df30672 : SUCCESS in 51m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd60f8c06fae4e16baf0516589704e65 : SUCCESS in 33m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0376b22f2c2a4137b736a9f8bf13c042 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad6fcc0ad3f94b4f97d3a94896b0ce30 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6db41dbffa2e4bdeb31ec07cc0fc4f4e : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5490396e0e444215bcb6d6b9fa7943ad : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/431b39e06f894180885ed27a6ac9fa00 : SUCCESS in 9m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fcdc68654cd64019b772032c443e6377 : SUCCESS in 22m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fcb3b2d51a494a179eb9e6dcc018d04c : SUCCESS in 26m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18691125364248e7a68d62925f6ac3bd : SUCCESS in 7m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/580ab5655cc4439cb08462a415cfc22d : SUCCESS in 18m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69434c0ca55a472b98948bcf1f6cca66 : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e523f0c5b8d641f6840a8b73f668f5fe : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65e702ada3834f3aa9e4dc77545706d0 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94e68f3cbe404dee9dfeaa7961f61fad : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3438758873684c39a997af9e699a4b80 : SUCCESS in 32m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30f816de7a844713bd97dce9669a363e : SUCCESS in 53m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96148c2f345943a5864b84ee3d4d8a86 : SUCCESS in 37m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0246480fa3d4ed284ac89f56e051b33 : SUCCESS in 26m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8059b96722a54a4ebf8d7f174a0930e5 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8dae1e5481c4416dbe36cdb903e6a29d : SUCCESS in 20m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ddfe0285f444b53958edac4af173c11 : SUCCESS in 39m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0be32399163247bf9027a7eda01d0f1e : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":6},{"id":"dbd41ee390257f2b9fb8b35feaa78957d686dc3a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 13:12:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d473b7370f1acaf08ef9c81eae81e85deb269409","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:20:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c16f7e388147698c4919c000351082\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd8cbf33c3164bd5999e72304e7a0870 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":7},{"id":"2296070cb50def168fed30d559d296dc52bef6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:38:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/826ed7b9c6bf43588bcf97255d84105f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5921107e9534885af2b437e6d162a95 : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/844cfeda9bb74ac2adfa9ce7ba48d040 : SUCCESS in 1h 05m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20bbd6aaa39141e18b8f88767c08bca7 : SUCCESS in 54m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7810a28a0daf4d718f436efa4f1e3405 : SUCCESS in 1h 04m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/303ce2456a2841ebb74e11ef3668fd41 : SUCCESS in 1h 08m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dae91d3d69a44a2e907c44c8a009ac2f : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b443883be53d4f45af7c7ce5cc60f849 : SUCCESS in 26m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5bd5b2e79c44776b1ed3b97f1993c0b : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0aef4ee72f4541ccb2afda7e4517e78d : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8f6ae8736e42d7a7dbe3bce45ab748 : SUCCESS in 19m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef9f01a0cca14b56be29e74126d030ab : SUCCESS in 24m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a044960433fc4673b3e89b4a340bed08 : SUCCESS in 12m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d7c3be1b5d1475e8deae5604331cfb1 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/15da635a00874a5284b197c51579d72f : SUCCESS in 13m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5fe6a4d455541ea8310b1c4f7b47278 : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c013eb5539e4290953fa3b786bcf672 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5e5d5f083d7410e9c6a12202a98926c : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/60ad97d318864482ad3780306f45664a : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/09fb38123972418e83a156471071d159 : SUCCESS in 52m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08c5825e6ccc40e08703a6fc3e076f2e : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/768d610186fc4112a0d5ed8bf2d405bf : SUCCESS in 50m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a3abdbeddf2491c893147331af7c878 : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09997d14451470899bd51bee9d2a2c9 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0884a9a9cd2646fa9a495af13bbe4754 : SUCCESS in 20m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac087c1000494a27899706c183977ceb : SUCCESS in 1h 17m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58f4811da3174b4d90a34bcdaed957a0 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":7},{"id":"eddff622d1e4e67f30b2c266bae624aecdd9bdd7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-19 19:58:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1e52408f81d49e3ac886bc9d499c4960c6dbb67b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:36:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a4bc4833ca4c21a80bf2a4ed42f160\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab90483cee9c44469f967a6d858d2a58 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":8},{"id":"c1d9a6abd54512c7003cc496ed2d85a7f7c9a064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 21:03:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4892434767c4e44979d86ab12ae1650\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6595890cf2cc4fa396996ef31ecb3a8b : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e054c941dbbc4a56978e8732e1c7d325 : SUCCESS in 54m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1764ee75cae94acda3004b96fa6e1974 : SUCCESS in 56m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9260d865ea0a427297849d9628f2b170 : SUCCESS in 42m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8582daa27fde4864924565d93721a36c : SUCCESS in 1h 02m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10ce0ebce2784f9a956dad165ef8b352 : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/035b9b8897be4fadb927758e154d8916 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/103d5c09fd03472a976ba0291ab800df : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6cd0545af0b4d2fa9fee239174703d9 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4bba8286d6f466b91f1215854cf158d : SUCCESS in 17m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/101b30f55f804f61bb112ae2cccd9940 : SUCCESS in 20m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d12a9157f0345de98bd8b203cc88f09 : SUCCESS in 24m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/957234ce15914f1785dfb43fe27ee631 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9eacaa548fe848ab93bb5645be7f4501 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a78a5a67616f48c1b2aae128cf740925 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/79f1252aa1a147e8a27807b9e4292c1e : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/420b8d999fdb4d4cbd6f59e1c05fdd20 : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f3934878cf14487a962955a7d1f2290 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb805f39a8bb46339af8a2df82aa32c4 : SUCCESS in 57m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24334a7f974345b9a46c9f8b7a35661f : SUCCESS in 17m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5466475dc3cb418d981112770f05cc2b : SUCCESS in 44m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5162bb5ad0e34bacae191a54f62a611b : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bedc695a54141cfb164497dd9e20aa9 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50e78be791084d548706cd0c6446faa9 : SUCCESS in 8m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2aadc04c1c045a3b9a4662c92463ff1 : SUCCESS in 40m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6aff20451e847f6b8aa4d4b62ed1742 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":8},{"id":"ad723959124da7d955809b351733a53cf4d00962","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-17 17:27:10.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/proxy_logging.py\n* test/unit/common/middleware/test_proxy_logging.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"56149d040ab8ce62ef295c3810c1c18327d0bee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 17:35:34.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f050f456e94d4dfb85e6c0e5fc01c155\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d92aacd1a4b248dea854726f276971a9 : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":9},{"id":"b6fe12a97d032775c065b8d5b1cc2f191b86cd52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-17 17:53:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4697ba5121e4b570dec969501aa851f057bbe8f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 18:31:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c79c26e2183408992e3f096f9913f26\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e529b4f401046418f80569bb6c7c92e : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":10},{"id":"7efa9b9c18ce12dcb3ec5b07f654bf40076b1e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 19:02:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c578e0c346a444f90450d570239cf95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970fd1dec1724b83949f88ebad0bc2d7 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/dab1edfee5d840a39cdb1db4dbc8d98f : SUCCESS in 56m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9519421fb5c4bbeb39f23532780e7cc : SUCCESS in 53m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7dda258158e451993d5911edbee8270 : SUCCESS in 40m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06c711d8f5f0448fb77ca38e57aba9da : SUCCESS in 1h 03m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c42e50f7ef9475198b1f2a4a66460ec : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/baeab0ad5d2a4b12912eb1a1fd42859c : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/407bd8d6bb754be48f00270b5070f679 : SUCCESS in 16m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1645c32dd86e4a909a304287164e1826 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0064cbbd421548bbb35d0d6a311b13f6 : SUCCESS in 11m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72e584fccdf843f49d71b6369f1edcf7 : SUCCESS in 21m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d69066ea3e084b19a82f2e22a867e2ec : SUCCESS in 16m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2cdab18575204413b2015d9fcdcc86cf : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/508e98f371aa4342bc2bdf988e109db6 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d91c5141520443c38eaeb9f91737f107 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b73c0b933d943739b5ca91db2ae4a82 : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/960de77b7a034356a64420584d5297fe : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a4a149203db423d8e269902a1743fb1 : SUCCESS in 9m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aad3cdf670b94a3ea74b5fb381f90883 : SUCCESS in 55m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/56514332ce324d0587891181cb24e9c9 : SUCCESS in 30m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95cb8e6a63da46f98123feab5c7cf58e : SUCCESS in 30m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a9a175c21ca4b9bbcdb8c16b6bc2060 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8e6ab347bb453ba9e6d8c2b8780c49 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a02b99e3d7e84f71a41c8886ff2f4f93 : SUCCESS in 14m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2503561c9a24b32a7a760cfa04a62f1 : SUCCESS in 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8956fbfa6c4a45468fa3774e533333e5 : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":10},{"id":"1c0a1d20db4c126fdbc9b358bcaf725660cc39a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-28 01:18:35.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"a075277e4b317b1276fcd10525138cfb202d28ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 02:05:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dea1e87ba34453fa79e05d27fcf9f55\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d33e4a854fb4ab69adfab8e60d5128a : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":11},{"id":"6597ae6203f081a4f76c8c9ffc9352e61975c336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 02:38:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e508b7f97041099df2aa017438ff88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10633315332a4f7a8704af60235fa3e6 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/bdf38e0e682540268f3a4cfb0e950cfb : SUCCESS in 53m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5a32be7e34041aab51c4a5c03259e27 : SUCCESS in 52m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b16b76b2eaa641c982bef38067386f02 : SUCCESS in 52m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0cd1a73101a4aaa85bd95cd28f73513 : SUCCESS in 1h 03m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02b5d967b216465588ac458eb0d6f43d : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e484c1507d04a40a035e815c3778910 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73cff2e6384f476c84f30c1be64b66e1 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11ae2b36ae5c4740a2a326eec8fcb394 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/38512389cb874b00aaae87e252e6d48f : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d39313d37d94f44ad1e582185c22628 : SUCCESS in 15m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26044b58609c4ae7a568c37c33c2595c : SUCCESS in 10m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5a7be26007854847adfe80f95057a0c7 : SUCCESS in 21m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e646872f6ee14795b70b38b5ae505e45 : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c70ad98ae16f46aeafee9eff96edfff9 : SUCCESS in 7m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4d2aa13164644538ae48441cacde3a7 : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a8717fd50f94c12abf6adc7cd561872 : SUCCESS in 5m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e4bf22e87f94ad0b507ca3a947cfbb9 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce48001a401f4221b23df2a6eaa47d05 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ee3e4edce504975b61a64c3c092426c : SUCCESS in 54m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb6364a845eb49c28bf546d129d75413 : SUCCESS in 28m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a8b036cc21b46069da1dd9658d3c2ed : SUCCESS in 46m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58f5a16a9b854e259365821a9386352e : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609b7114743e40c0837dc539200c7f2f : SUCCESS in 2m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c13da57695849a08d7750336b14d733 : SUCCESS in 15m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c7315d3020e44d68290837d64b20110 : SUCCESS in 37m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d34aa34a792449f58aa4c3b12e6f993e : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":11},{"id":"c54a20fc89bd3e9b552b51dcac6ad68ce3165f70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-29 13:17:04.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"779d641830c3d413da75ae54c60b46240f5aa082","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:09:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03be1fe010334546aaf70bf7346d9337\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de5e4c087e4d44ee91ac084c75b10f5d : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":12},{"id":"797c2210ef602fa144d5ce981ff8198513b6481f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 14:32:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e1a95c6883540c2bfaacefbfe0d4387\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3849bcdb784444b0dbaa4c1ea04d84 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/3b57520c336346f6933eef25967f90be : SUCCESS in 55m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/245e221af0c94306ac092a751900cbb1 : SUCCESS in 26m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0c9b56ee4694fc8ae8de95e9fb1de5e : SUCCESS in 29m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/215a990fdff14060b7f7f275d7f0f445 : SUCCESS in 49m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdfeca0801524cc1a6902c4dc3e87b10 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13b769a1b90945019a1e692162abe09d : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d138d87aa58346798f1db51d0549bf42 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c2bec9d664bb4f7abca9dbb1b986dbfd : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e1020a0e6854fbc818fbd387969537a : SUCCESS in 19m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b92fd15c751749a9b6b746ce69c060b6 : SUCCESS in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1331e6aa17864cce815c6c64b6f9f4c6 : SUCCESS in 9m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2b69c40abc25457f8d79990b7bda8a66 : SUCCESS in 23m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cecf6de697874d49b0c68e96315aae81 : SUCCESS in 11m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b375dea782284bb8a6824c350bae1fbd : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/affffb92f8654e299dcfabf9c5d82ace : SUCCESS in 7m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5912e2ead26d4bfcb447a51ffbeb4153 : SUCCESS in 6m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7eba4853306946f89acf1805fdc82a6f : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8efd15ce84354872ad1b875eee21eafe : SUCCESS in 3m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c33c057bbf34213b8d309788170c101 : SUCCESS in 53m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30c346663c5e41efbaf2cab09fc1ba82 : SUCCESS in 45m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/074657215afe4c3ca36039c49cf26490 : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eab499e37ba94aeabbe10eefb26eefdf : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e76cf6b1774d7986a021d21f221205 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33cc2f5cc872411d99bd2934ceeb6b09 : SUCCESS in 19m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74db38d0c0494ddd8376c40540af83c8 : SUCCESS in 59m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b754c31150dc4079b2a4f53313fab725 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":12},{"id":"7c1edf07452d19b36b7e2745c1891901c8e6624d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-30 15:18:43.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"88c3e9bc487b0b3aab76f52e3ca8ba365f498ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 16:00:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e455cd4f69c545f095ddc3420bc1e114\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe4701856e284cc8899dc655cea12d4e : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":13},{"id":"a34da2d690160aed60783f9cb1fb485c2ab6ba70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 16:26:32.000000000","message":"Patch Set 13: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5bff68a4d8641a087eaa81c1389e374\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b9a380dc7a48ba9bcf241ed1da333b : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/55d8585d980e4a6b9f74f7c809801f52 : SUCCESS in 57m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94186826e38242b6b3278bc7565beb8f : SUCCESS in 53m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/946961bec15a4b059ae8ece2bc1fb4fe : SUCCESS in 31m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b3cf162eb154999aced792124800ab4 : SUCCESS in 1h 01m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0150c4da5d9747c8b5f5f0e9624fabd7 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/594a81b3fbe94126b93708369df02a08 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d8b1692209e4d8abb2fdcd1778e3df2 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3121fdcdd31d46b58112216f5084c75e : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/889134b466a041fdac547a0bd350c4bb : SUCCESS in 15m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e6f77249f7e4be192543315f41ecbd1 : SUCCESS in 20m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/888fbb7639a34a6486aa5266d75e395a : SUCCESS in 26m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d2a8630d304420eb60d3c8bcd8bc611 : SUCCESS in 27m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1f26854cb9447ea85e8b3fc522d157f : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/67fc8f14d36b4bf0906ce8e2f18341d1 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/548a7161149d4ee4a54cfc2fe057566a : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4f22055cdb24081acb73ebcd0b25fa8 : SUCCESS in 4m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8e360052fd9f42f2a2f7070efdce2ece : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/facbeddc98b84cfea85a3f72e088634a : SUCCESS in 5m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a44d0eb25d5444298958a4eb85367c0 : SUCCESS in 57m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6600df627f504415983fd6286377a18e : SUCCESS in 57m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22cc7c0fa8b44a15913b2ed55488b9f0 : SUCCESS in 30m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f322a82f8ad4cc3a21270da2606698a : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62d35394baf54c8592f90417c78c6a51 : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3e0e1f5b2774a209f9c108b54d0b2bc : SUCCESS in 23m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14bd4bc0c3294848949c4651509b2603 : SUCCESS in 53m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/643e5e1ecd1646ffbb63f5023c3cf17f : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":13},{"id":"b2bc8592ca6d72befcd80fccecf60ab943162f61","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-30 17:48:08.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"64095ef7ddacc4cc8bee26adf7a3b3c4f56f7913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:03:36.000000000","message":"Patch Set 14: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/829e42007b9c4a3597fd4ff9b5a02559\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b9db0afc6642e1a0ada307eee6efd7 : SUCCESS in 4m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/bbef7348263d4329a3ad1d5a55602b9b : SUCCESS in 59m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d94f59672862431b83e4c6fecc36e29c : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c686318953d24ee1a00e60b00652caa6 : SUCCESS in 51m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8ac5272250347b698b93c08b0bc7e0a : SUCCESS in 31m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f969dd829b4b4a23b8ab5203586cd3af : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b4c8c4f459748c29ff18c06d2ad2bc4 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a9a96c6326f4687a37714ce26f3a235 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ccae0c1d0b2479193f2e9c18edbf3aa : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/810dde89fb3b484abb2a3cd5f31610b6 : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd076d2c5ede4f35baf42e6f004c43bf : SUCCESS in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/065380aeabce4673bfebf2aaa5ecbc91 : SUCCESS in 31m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b7067bdd02324d98adc8be40a757ca3b : SUCCESS in 22m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7615bed1e8e9462fb164f9e6a92671a1 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a83a87a6ef784c55824b47515f433a57 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/41a74a725be54168b632680018b25aac : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb3afcdf8f73493097654fbe9c26381f : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d6c38eac9c84363ae020ba49e63c931 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0455941414544eff879e2bec7d72b025 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5cad17df1f042d78ba19feae6c11754 : SUCCESS in 52m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4ccf9e9c7664d1cac6d6050cf91b5bb : SUCCESS in 53m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b2c46ce6e6b4e73a103e115bbec1820 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9462b2bd2aa4fcbb61daa30ac489b01 : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4db52df7842a449d887e9a88bdf2101e : FAILURE in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/188d73ea46be4990aea808d05f8548e1 : SUCCESS in 20m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d2012175bb24ebc8356657451de7cab : SUCCESS in 45m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04be8cef81794b8983c2d35c3d46cd68 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":14},{"id":"9e64c0e0ab767eef5146d5f6497f371aee205ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:59:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fa68a327e145a08dd2d9cc46839645\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7ef173c06ac40e59b8244909202e5e7 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":14},{"id":"88c0f602bad6646ad76398b130fd6a5647489d44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-05 12:20:12.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"259f38e3fad97395782529a0a7062ee2e6eb0dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:09:25.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca5e2ba8480042ef86d23665e56159e7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec6720927f7f4ac79605c13290e14128 : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":15},{"id":"927abf406d757fc17ee2eb51bdcefba6b717a724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:26:43.000000000","message":"Patch Set 15: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfc0d2221fc047d696e328670427730d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d072a9755ff140aeb8fdeb588e3d9d0c : SUCCESS in 4m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e553ed1d714b4e4ca62e1ee3461a638b : SUCCESS in 30m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d3bca63f7154972b771dae312d72cac : SUCCESS in 57m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b7b2fc4885f444e98174c52f8996e37d : SUCCESS in 1h 05m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/775f9c73c07d4cfbbadbffe0b2882b53 : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0691e63bbef74c5fbb4ef020a6c2d852 : SUCCESS in 10m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0925e1a8cca34add92c880d9925558da : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7c310d5bc2f481881c80f874c8665fd : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62270b3fb48646baad30a030d18af0f3 : SUCCESS in 4m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14ef05e48da44606837a4084fd0070e8 : SUCCESS in 7m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74cd41ac89414e21913955d2da8f97ab : SUCCESS in 20m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e3b471eb9f0446a2aadcb14b87c6b96d : SUCCESS in 27m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8e1e64780bf546b3984c61ffcdfcd009 : SUCCESS in 21m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8347f3a2a8d3470faf3d5354ead80593 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44a63dc0a84c44b4a81bb67e60f38e90 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26d52aa4524d4b85b675d798093dfc91 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37beec5199624e969fc118c45069425b : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f8d28bca584499d94aa94c13cf3bb90 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0b89f9016c64c499c2037c6ae35a947 : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/858f0749247a4f41a1156105843381dd : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08c8513809934ea181e8690d099aa3ce : SUCCESS in 31m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f5f86a2989f480bb6aded775034cce5 : SUCCESS in 18m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00de683e341c4a0c89593d441692a8dc : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40a8e2afec341f99bc99e01afe71055 : FAILURE in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa5df00be5104d46b1f4c3637c9e070c : SUCCESS in 8m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79bf00c5b52e4ea687ea30396fd78144 : SUCCESS in 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cb1d2ca18564d00b06ead099a0d8a2c : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":15},{"id":"49c82d36fc3f902c9c5a04d89b31b9cb41c0ec27","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-05 16:16:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":15},{"id":"8e2fd97e844318c606c801c63713f678e0be5a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 18:57:46.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"816f3217a6cca55fcd006739a79296223728f742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 19:55:46.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cfe3c625fe84428946cb0d7040e0bc8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3a37d8b56d8492a9c28cbba0712d492 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":16},{"id":"a9111c40b7853d67943f6aafe13649bbfe37a1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 19:56:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9dc461362e44b4929f8772c9a819e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52646c82f524954b986428797aa5e0b : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/efbf6b2ff850416286a50a404c813d44 : SUCCESS in 51m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34acbbdf58344e7097e18e2c0dcd18e9 : SUCCESS in 55m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26549e5678bb4cce88596cb2472bcbec : SUCCESS in 51m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41a32b875e99472084bf9f662b5316b5 : SUCCESS in 36m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c800a1dd249c44a0b6b4b3cda5b70d18 : SUCCESS in 10m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19657ff46e2e4561b330a238d4527f9a : SUCCESS in 11m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5870eb3bf8046248f48c6b7e8bfdf76 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4abfe0a6cb784445ad2e63c3e45569cb : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b92d59935b24190a22ba4e1e0b7486b : SUCCESS in 8m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a44df37bd1df421c9a433280ee2616c6 : SUCCESS in 10m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9c3c78d21d34bcfa44a98ba9e5ef25f : SUCCESS in 33m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e10c7072611f47359c289cfa7d7d70f0 : SUCCESS in 16m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9ebfaf6db5045099bea8e1e6ec8733f : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/164bb789fbde422ebaa7c00fb89363bf : SUCCESS in 8m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69806c5bcda147c59922c7f7065ce184 : SUCCESS in 7m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a91dbd4441e4817b2aa95c78daa77d9 : SUCCESS in 4m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d89b60db4e14b5da6345376657a482f : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/008e4a1988984f02a2fdcd695e1030d9 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb9c9a3acf7e4696b365608533cc2a94 : SUCCESS in 56m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f95c2b45b03c4d53a314ce0264e3b4ab : SUCCESS in 32m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36b23bb8c0884aa5808958d89b9b150a : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a5fa975220c4f38bf1504f05a2189b6 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed7984bfb2e403087b3be0a1a6a5ed7 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec8355db33d847678c737190804ebe86 : SUCCESS in 20m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/081f5375f47b496ca6b5bc9d485dfbdd : SUCCESS in 42m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/281654f5adcc47ce9fad1e8da80eee2f : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":16},{"id":"fa2fa5747df790a222f3b6b3a13571347686dade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-06 20:54:19.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"fbbf4ed81adb31be4ae35aa10c50e415952eee1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 21:31:43.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735e8684b53d42ada0e0e9350268b6fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94cb4ecbb8fd44af94fba6d60dc8635d : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":17},{"id":"35aecc8af9ac3395971c43a177a86f163fcb30a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 21:52:17.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee70381009ea4b2d8bf913d78a4581cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c3ef7aceab473aa05b2836f14dc7c1 : SUCCESS in 5m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/576d91e1b63e48399c16bbab7912629d : SUCCESS in 27m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e467c73a26c94489852562cd4e3af226 : SUCCESS in 42m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43ed1aa5bcf64d97b790288b2d7cd273 : SUCCESS in 28m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1151b9cf5f71484d812f0e7b8678c064 : SUCCESS in 47m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5975a23b6364c0fbf43e38da73abfbf : SUCCESS in 9m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bba6bb76e6554f5c84d15f967068a0f7 : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36243ac2b610468480f2383f843d42e0 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbe54f462572455995636e794d152fc1 : SUCCESS in 3m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd626b3c27c84d38bd0f6c2578737103 : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e69538887ca4ad988dc2004a51f22d5 : SUCCESS in 13m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c2c039566d254e27a8bc8117f7b9409d : SUCCESS in 16m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/95e04ff1d335405b8d3749292f9fc554 : SUCCESS in 29m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/22a52a7eebe946329d5f51281f8876a8 : SUCCESS in 7m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd17ac492b174c79938f4c92895aac98 : SUCCESS in 13m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7ecd1d9b688445ff970e4eaa9db7fa21 : SUCCESS in 8m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc498ec4401f4fd39c9f4d7aa569bd01 : SUCCESS in 3m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6b3efba16d3408bbff04b48d4973d3c : SUCCESS in 14m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/55c9d71f96434a16842ad839cfcfeb27 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ab6bdfa10b34c70b9be47c9b7ef6e60 : SUCCESS in 57m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e7384f5095c4cb8a497848368a8f233 : SUCCESS in 22m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd82c87d95dd49878cec35c81958326b : SUCCESS in 43m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4ccb8ae22c64def9f81a1062cc08b5b : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282c90488e1d4dd3a88328696718b8bf : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd26078b1280461b89559628907d65b0 : SUCCESS in 14m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20cbb897002a42d59c3f78b6771ea034 : SUCCESS in 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c5b2c17d9514bce896292708065d4a3 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":17},{"id":"918dab674ae4cff3ddef63fc7b84f5c94983574c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-07 13:30:40.000000000","message":"Patch Set 17: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":17},{"id":"08babd253acfee54eb392c84b795f71437e26fa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-08 19:56:14.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"6ba87e01a83ed66b4b0d8db76ae81885b315c531","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-08 19:56:32.000000000","message":"Patch Set 17:\n\n(19 comments)","accounts_in_message":[],"_revision_number":17},{"id":"31a7fafd4c741eb5e80ff26c91915f1e4492028a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 20:30:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47d1c6c81ba9464f9b603fa4725341e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a24445e318c4bdca3b4b6985389a430 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":18},{"id":"8a7b7568036f5738904b6b441249db691033c06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 21:06:17.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db11a42f02c243fb8c649504480e90ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a65ab240a7c474a9f72aafa24207198 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/074e6f57ebaa451987fd8d5eb4350759 : SUCCESS in 51m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76936d68f9224653a6690c2b73e2c8d7 : SUCCESS in 43m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58860fb9c73a43f8b73721ff3caf2466 : SUCCESS in 40m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5fbcc42452b4004ad03e0a27af64de3 : SUCCESS in 57m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88f6446c7ce04578b3586e84e57c2129 : SUCCESS in 24m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/542de4bbd3384dcea3eb3c0dd8155745 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcdc3ba6b8f04d8ea8acbd46ffeb938b : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aee5893f5e474d6b9f58dff47e47bb36 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef7ddb5553644d4dae7af4992d6d4d2f : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ee71e5bc86e42858efba17a35ce0344 : SUCCESS in 16m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/800148fcafc74bb8a6eb32cf40d2e046 : SUCCESS in 21m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/79e58ebb6dfc4e008a8489c2d7197ba4 : SUCCESS in 25m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b71af63e62e49c9833765ac70adf5d7 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97af59920de14a6aab85e3c03a0893f0 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/203dcfcd327e4ff8a9c09b8cc3738a20 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd627d79cd3048fdb2cf25d3e895fe4e : SUCCESS in 7m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89ae556e2a1e4789b2d33c159e288550 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16a47bf0877b405dbbcec4319618c4c9 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2c1f8b74f824bd9bf55aa7a01de9349 : SUCCESS in 1h 00m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b124808f29d41358a44c868546331fb : SUCCESS in 38m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6435a7e4d89469598b4030ea052c785 : SUCCESS in 45m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d30af14325e4b549d66670d89de8c5d : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c75e6868bdc419ebfcf15400693cd3d : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/830418dda3264a0ebe50c78a8782accd : SUCCESS in 20m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/564ac3da16134061b8e2680333afa54c : SUCCESS in 47m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5b6532077a14d21a676712d51ea9e96 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":18},{"id":"a43e74b999366daab089e8bb97ed9594e46017fb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-12 09:19:28.000000000","message":"Patch Set 18: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"c968cb3dca22247f012295b42cf3c0465f560167","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 12:52:30.000000000","message":"Patch Set 18:\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"1197666de8b4756f3e6745cbbeec52366c4e1a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 14:20:50.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"806230644ef1cad314831de596cb0972318f3f34","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 14:23:57.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"73ea803b339da57e2d4b7e9926fef820342d5b96","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-13 15:17:47.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3143ba7488de88e4d7d479fb6d91644c69d16eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:19:46.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c79dc5e58ad40d7bd6bc8ce9509d141\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cfc65a76ce1146e48286a6387536090b : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":19},{"id":"28c204d7798677a902413398ea0399293d137afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:23:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d9c6f2f3724f9482ca85f40a62e7d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a982ce6a3a0e4b5e93b198ffa5bb304b : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/76b285e078104af18ed817e62d29bb4b : SUCCESS in 47m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b654259c44142f3a9e820c4a70ecd98 : SUCCESS in 49m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7deccf215be3478891678e7e0745ba54 : SUCCESS in 43m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad25b468511d44388684687bd0da9b57 : SUCCESS in 54m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ecae79614964349a57b9e5cca0df928 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c61373c36f243a8be0540f972a39fe2 : SUCCESS in 23m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1197ad6527504f91879fc680e6dc79a5 : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59b665bfc2f84291b46d872228ccf8fb : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae4da13d4fd47f4bb0de44921ba7b6e : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb6be1d60642460da492f14cefce3005 : SUCCESS in 22m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ad7b6761690a403e84751daddff8082c : SUCCESS in 19m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1350e7f75f6a4d3fa5f59105a0fdbc09 : SUCCESS in 20m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5817c94cc7b6440a890fdadb117b4ebb : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/860a7717cda3418584096717c08029ff : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/925024b86aaa4fbfbe98b97218496e61 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57b0cc6fd6f1449c983eed4fe0c9fa9c : SUCCESS in 3m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdf35d13814643208a713b3c8578fd82 : SUCCESS in 9m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/37e68f8bae69447eaed90c2aaa0a5cc2 : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/46b43d3500c34a49a839d8be7e10cefd : SUCCESS in 56m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/615465b8ea60415695deb4814d11c565 : SUCCESS in 17m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dc495a3d6854196a30fb95d5228268e : SUCCESS in 44m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba9c1821d1be4cfbbdef187a36da4ca5 : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed00b932a234037ada91649ec4a4610 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d53cc30296842378af37675acb1d372 : SUCCESS in 20m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c7b9be2c7e40ca893eefc170a58991 : SUCCESS in 52m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41450b55cada4442b8eb44375cd7bd49 : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":19},{"id":"67f2910afd5228d5d4de3149c6f0198c33a63f16","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-13 17:14:33.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2f7cce175a4401e16b94f42b6dcbe47b5d9d84b7","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 02:51:59.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"b6633965f21dec476f451a90c98693c2c911cb83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 03:01:48.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"1d0868874b9c912dcb620c9b5214ce08565902ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:58:44.000000000","message":"Patch Set 20: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f893fbabf2d34341bb0793cf16888369\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88165835c0e48cf8454dc5d9100beba : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/19d8deeda826445caa1274b937e26b8e : SUCCESS in 33m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b75a10702b0486fad02c091dae8c318 : SUCCESS in 35m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/549ccf2802044c6b8f006040be4cb2ec : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7255e0c42401471ab7879ab0fe59846f : SUCCESS in 32m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e854978773c34a82951f61840436cf5d : SUCCESS in 23m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a6bf9439e054f9fbbcf5510f866aa8d : SUCCESS in 23m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4a95c89fde945fc86c1f95b9b21aba8 : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4120fc800ac64d8892e58650b8e43ec5 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9af2dbf03c9047dbaae4d7f9984a3702 : SUCCESS in 21m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f767e4aa5e554e788125eb5f161c6acd : SUCCESS in 21m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b1e0e9f046e4452dab554a35d391744b : SUCCESS in 26m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a966fb64791d4f5792fe8e9d9c4e8278 : SUCCESS in 27m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b215b4b40474953ae0ae134aefb5df5 : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e607a59d0de048b397dbdb36b0d83e51 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0a9c8d3cc7ed46b98a174c98bd04dee9 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b1a78e7d5aa4bc2819ccdef55fdf545 : SUCCESS in 5m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f85d706f17ab423984e3043a63413ef0 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/573308e83878498a9b11298c1101d720 : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1260ba7ce98845d4b7fb82c84043b376 : SUCCESS in 47m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8791b690d145444c9ac9b07c3419f021 : SUCCESS in 46m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa8da803c6434317adda7738054d2e94 : SUCCESS in 31m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0540816c82b4359bb14d326d565ea9b : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5db4069c8034512a9a3af9c8c3173d5 : FAILURE in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57d1226e69c441c0aba3286105563723 : SUCCESS in 16m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/220278d6ac3d4f80997506c934e99f5e : SUCCESS in 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0371047bca274b35a11add40ca6f231f : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":20},{"id":"0ebf41b4fd862d1672f2e62d163230840fd5f510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:08:40.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa07aa6f48b24c5d87195a4aaac5359d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99241b5f88fe4617995ae94d58329807 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":20},{"id":"3026fde2280aea87c18c7ca15698bd59f15fd303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 12:39:32.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"65f49e0c501b63b00ca271dfbbc5d30ee01edf9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:55:01.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a05e0cb92ba49f5bbd43efe15dfac27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3623fef2b0a4574abb59d8f423ebd8c : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":21},{"id":"a4249010a2d0f147f0f4f87948eab3a7df216911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:02:57.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea7fe303cd0e427fbf73ad02fd2cd167\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b224a43e1d34f1ab15b67427e5a824f : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/1c66dd03e0b34f4ab758c68019957e61 : SUCCESS in 56m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7f7ada767f24d0c9c713dca4fa5e1db : SUCCESS in 1h 01m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94bf8845036242b5ba7ef14c9eef92c9 : SUCCESS in 1h 02m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f6ae87dc2524cb1ace1a2a8fbde15c4 : SUCCESS in 57m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cd67b6e14a24f76b96a65a771b37ff2 : SUCCESS in 23m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ed8778742014f4b8b3bc027773823e2 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/362f931118d9470ea1d726ddbb343f0b : SUCCESS in 25m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77be8304ccbb491795e77a4cc631d212 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba5f1ddeeaf24d968f74a98ea31d74eb : SUCCESS in 25m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b23a7f7fa2024f50b73af3af408f4b4c : SUCCESS in 21m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7c93fd0a417d4eb9b1ed165e3b4a155f : SUCCESS in 12m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d1ea2b0143df4cbf976f5a70539ec326 : SUCCESS in 25m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/87e6fa5a3093401bb820fc4c6aff52d7 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d74c705df764454854fb519be34c738 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a115725a175d43d283bcb6f1894f137d : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03615a61f32e4e438a6e49865e2a8c97 : SUCCESS in 7m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98ff8ff0e6614d9b89583803f83554be : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2842c700816045adb18b2e939aa66547 : SUCCESS in 7m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/92f9029ea8984ec98dc1aa05fa64c243 : SUCCESS in 58m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfabc73e9789423ea77ae70b616e7a80 : SUCCESS in 48m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc7b5a4aa716418384dbeacf7b66f458 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ac7a877022a447dbdacc0ad8799ce92 : SUCCESS in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614d71e578434e59854e2025da8d20fc : SUCCESS in 2m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b29185c95854f01984e448251371b5d : SUCCESS in 20m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7818a63d6cae4168818be15c6c03c201 : SUCCESS in 44m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/812d143d7bde43c58ebf1b5ba79e61b7 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":21},{"id":"851c33d3cbd90c7bcd111803710ff0af1b855e45","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-21 11:30:15.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"246efcf56ab6a54079adbe7b419a124a3437c906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-21 20:27:52.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"7f4b992bb84ab88d9cbc47af43e4e7f8fe9f248e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:07:13.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95d8f0bfa1144a26afb09506f8205238\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c1ad02fb3a14d259a919075084be507 : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":22},{"id":"c2f82f5fa18f4c324b4ff3b4e0daa43cc99d086c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:45:14.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f05f9cb6d840a888c7ca30f64bbf4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c9edae54d44c02b5d48dcdb794eccc : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3b463762000d4d50b2fc87ed2eba1356 : SUCCESS in 51m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d27409eda094860a24427a1133efb1a : SUCCESS in 53m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49ad37fcdcd34660a5d1a2de0352663e : SUCCESS in 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42e973271010469eaaa54d4ee195f50f : SUCCESS in 57m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e3f4072278412aaadb02e6c5e702d5 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ef39f6c678247958a18c9b0dc93ea42 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4e5d455632347b3b81317f5c42ff2ef : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69cdb090edc04dcbb4917a8b59267f63 : SUCCESS in 3m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec24506a197b4de580a3d12e02199a8e : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67b1b6c32a2a4b04b3b096a7f1e2f21b : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0679a0f14cba4ad19a10969c601a5ec9 : SUCCESS in 12m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cba19ec1ba8b4b0193074fc4f2116dda : SUCCESS in 27m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/254810c4eecd4f8b8e8f19cc50249773 : SUCCESS in 22m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/17c7937325d04a92849edfa63971dcfd : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9e18c2a2a86047918625f5abed02de7a : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/935dec56808f4421a306af5547a92c57 : SUCCESS in 7m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/037244caa755470f9b8f565585dc824c : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3731db0d7ad491497ad0ca40a644f8b : SUCCESS in 10m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd241b86ad7647ab84b631623adc0e51 : FAILURE in 1h 01m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20ae218d1ee84a9ea4855bc1a4279871 : SUCCESS in 43m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/341000ff535340899091ab154ea79103 : SUCCESS in 47m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66727acb1bf54b31bada00a770720ea8 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1531915b0fc4162a5aa5e7c186219f2 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7334900cc0e490da37320f72fb0de16 : SUCCESS in 28m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/112f7976e5d14d7eabb776848f4275d7 : SUCCESS in 50m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2ce4a206a124e978c7f807e760820c1 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":22},{"id":"be939a27482a442120ef3180de669d422a5add97","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-22 16:42:43.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"1caa16669e06296a9183ff85c98f3bbc48a9646a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:08:52.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa07190688c429e80ec76650b4fb13a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bedd787c69f34c8f82f81b28e8afe851 : FAILURE in 50m 51s","accounts_in_message":[],"_revision_number":22},{"id":"a6ec9f3dff91577f5b1629a7a2f567049284ec29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:11:51.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b129c6e33d48ea9a637323475059f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33173fadd4844429440a5af073bd141 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c182a9bebc6048cf87374ec463483e91 : SUCCESS in 53m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4db4012be24d4fb19580107123a75122 : SUCCESS in 1h 02m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/363a7d7d34924953897e798c68a4830b : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a2291abf1f84219a78fdb0ea011c2f8 : SUCCESS in 56m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6982a2d850a4fc2bdc268990c8d3c9c : SUCCESS in 11m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1373df67a7644f9a894115c229134de : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98e5cf3ffa2040b292320c9fda25b6bf : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96aa48b7d0ff458e81ecb6ad00c4ef6e : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5522342cf1149ebb556d7a53ddb1bd9 : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d888d6c198e4b75a4f388ac79b7feda : SUCCESS in 18m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b5dedfea88b74ee1b397782d896c6245 : SUCCESS in 21m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0634f704a1074a3eb41a64997befc4be : SUCCESS in 29m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a791709a9bfd402a8b9ea96c82f9bd25 : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ffb66d915d24bd4a7bc85a4498b9577 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/06e0d85f37b048d69617b31035b777b3 : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d605b814d5a444d9a0a4f55e8a617ee : SUCCESS in 5m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8f7d526e0b14d0fb5685a56325ac815 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c163f05d1424d28b6cafd45a3530210 : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc01e4d0167f413291588f54ff94d8fd : SUCCESS in 1h 07m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6a464b572b84a09bdb52a9f2866b1be : SUCCESS in 33m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5cd3d8fd225426383b3224643393f57 : SUCCESS in 39m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74e56cafa7964eec9a29f02fffc2634f : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76360548b0ef47479319bd477b4a4c38 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2efbea2a47ea4fa59bffadbb461bf7f5 : SUCCESS in 22m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5da52b996dd240b5bd6ad73bd993655b : SUCCESS in 56m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5049c69f0b347aeb211345e4983d0d7 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":22},{"id":"4fa35133f97e88c2ff126da9982c7aee0536828f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-09 19:40:42.000000000","message":"Patch Set 23: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"0dc6be744d18c12eae0e0afbd1ecc8bf2acd1eb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 20:51:45.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f218b011583142ed97416f30e3cf052c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6775356e52ec447892189d3cd0089a1f : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":23},{"id":"03af66b061a5b3330e3a6bab5e497fc542d6e2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 21:05:41.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513d32f5462a4876aeddaed4c743c1e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1922bea5b54cdfa6e7a20301b6dcea : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d17f5bcc9f894de8901e702f502eb0da : SUCCESS in 1h 05m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/934db49570ac48ab81b10d1526496708 : SUCCESS in 1h 08m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3ed1ad782564f5c8d6908d24cbe3ba6 : SUCCESS in 49m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bf895a3e1e54e51980ff0a85c0c3fe5 : SUCCESS in 1h 01m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c46b70080c8d4226a2905ee032ddac4a : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a6d9b05aa3f43bea694d7aa3c28b2b8 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20fee6f885df49df89ab1b443601ca8b : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea724525b8b244deb606ecc40420c8ba : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dad50c6f5a0049e281367a61e1eddd7f : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66fadd2c16bc4f9e9fb3382ad345c82f : SUCCESS in 25m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae93a022c82140a0a0caf46494429be4 : SUCCESS in 27m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3d1f8c7c6acf48cfac00c7768c28588d : SUCCESS in 29m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57194b179501468b9d406f204aed02fc : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/13cdca30aea640a1af153f961a46e1d9 : SUCCESS in 13m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5fcd02501914dd5879179a9990d40ef : SUCCESS in 14m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29ac331c240f491591ab1d8708ebfd32 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c797bb44bb64536bcb509cc3fb66a00 : SUCCESS in 13m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ff1a34332ca41979528f351590742c2 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ad132bf77d24f63aeb5f89b4c722fc0 : SUCCESS in 54m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ae69b3246a1449f82c03334e27d56d7 : SUCCESS in 50m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86cc9f5c8dfd491f80092f22d16c86ce : SUCCESS in 55m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0e936fc16a64d668a2160bfac544b2c : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2714faf9494484a55a0cfabb7ec3cc : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/015eec031ba846e9a3558a38a235ac22 : SUCCESS in 19m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f39975c3aa2649f6830fab0d8a029bef : SUCCESS in 52m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c7d0fea0dcf4fbaab41a6d0c6409edc : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":23},{"id":"00b629361614fc13766cf8ff51d8359ed75fcf92","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 15:48:46.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/s3api/s3request.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"62959e81026cb863bbe3a612429e6039a0de3d73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 17:07:12.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/504fa03fcb734e7491ca61b6ca2819e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0179df775d324c5f88b1a2870cd64b5c : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":24},{"id":"8e12daefe6caec664814594a5395d060ddb33b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 17:51:45.000000000","message":"Patch Set 24: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f71c2845434c57852bca0d5200d074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbca6e14233d412fb3b84604e3bd5794 : FAILURE in 6m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/5b0a9d323d764827a8ff6873ebe6998b : SUCCESS in 57m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20536746fdf240af91af2fc2862e76a9 : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f36b53cc3ca243138c2105b403829ac2 : SUCCESS in 45m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8e9843b15db48ffa2d2a70b12e0e474 : SUCCESS in 1h 15m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0426e9c9fa7e4528be271c5a90bcae57 : FAILURE in 6m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecc64950ab4640758abb3e7376ccf51c : FAILURE in 8m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2bba0c2b38847e2bdf05d453e20caaf : FAILURE in 7m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30b812c1d3e44683ba42549a5b0acb77 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5e33172b20b344cba30b4dddf7006bbf : FAILURE in 4m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efed2cd1c1fe4b9981b2d352d1e7306f : FAILURE in 6m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/589ad701672d43eb97a9406e4a61d048 : FAILURE in 7m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3c4dba97446940c6a09f56f5359f53db : FAILURE in 8m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d04106248af4125bece8a2f09c6c994 : FAILURE in 4m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c33fda3899c54d4fa7a9af4e3175d36e : FAILURE in 3m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c6ec552a2d6f4ec7985d4e10c4e187fa : FAILURE in 3m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35a1f4c6606040f786713e2e25747271 : FAILURE in 5m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa7b2d1098304ef290ee5f72e2b148e2 : RETRY_LIMIT in 7m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16a3d3e8e3414432be4a892750d76a11 : RETRY_LIMIT in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fd502d5e5884cd9b2219111fc1e3532 : TIMED_OUT in 2h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/654e79ba77574f2b846193f4cff6533d : RETRY_LIMIT in 15m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f50730d61d774f3681678e01410b4663 : RETRY_LIMIT in 15m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4702755a895d4151b7271c963f83d44a : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065b43457d334c0cae82e906d17993ce : FAILURE in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5978830c34094381bd0fd772e4e0022e : SUCCESS in 16m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7cd6b960cdb47b9bddad4efdba0f33d : SUCCESS in 43m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18551bb8107e449e9a8b6e57ea08fd02 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":24},{"id":"80adbee13e95275269d366c55961a0162d3ee859","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-16 20:46:50.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"37399325dc63dc34b018e5d2acccbd07571cb3b4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-07-16 20:59:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":25},{"id":"414063944c91aeac8aadc2269422b6c722c46642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:00:38.000000000","message":"Patch Set 25: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bb0eee31931457aa1e1ad156806020e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f8767c3c504211ae83d8eba8d6a379 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8fbb39e7d78f463398d0408e3b9d6dd4 : SUCCESS in 27m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/628bbd078a054731b5fbf582c61bafe6 : SUCCESS in 57m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8ce104eafc14169a65715f841d13625 : SUCCESS in 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4364982a8cb2459dbc416dcebf26e013 : SUCCESS in 32m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c261314a9424472f953814eb1c1db242 : SUCCESS in 10m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc4caa32fbd2447482c801fb213ca8c2 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1656df7c778742c0b14bb1e7e4bc180c : SUCCESS in 22m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b462b02784f84932a92e91c432b18856 : SUCCESS in 4m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17ad04c009a845f6a30285e08a7e42b1 : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/416de98978404c878b0955b388d432d5 : SUCCESS in 14m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3cec0693e0fe475088a437a35915f751 : SUCCESS in 19m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f851c9826f54a65959434a07156db4e : SUCCESS in 10m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51209d666d124f138c9eb9957204f27e : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e023923709664066aee108505cc2a3a6 : SUCCESS in 15m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/32477c4c365845db912eea323fb3bbc5 : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/34eb462bcc2047e8be0679617d3b1a46 : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b4eb7df536848ab9127bb92e4c7cb11 : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/24858f65609f4d1f96fd4a2d777f7cfd : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad86e41548ca4fa8992fba792fcd24b7 : SUCCESS in 46m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c2a84335e034f33988644ed132e71d6 : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df543a3c8d8d4eb99011bd386ab779ae : SUCCESS in 51m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/692a92e03b50434ba991e8c7dc506d3a : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b160ed3fd24502866b3d9796b4b0c2 : FAILURE in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7d7cbbbc8fd44eeab339a306df0bbb7 : SUCCESS in 8m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8ff41cdf234308b4e2b543f82d52fc : SUCCESS in 47m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea99ac682bc64174abcb3394f614ba1e : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":25},{"id":"5ee760c7c2439b441f1e208e8817202e250cfbbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 23:21:46.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73d9461317a4448b4cf9893bef8164c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/398e7953bbf549f9a3d3b3e4052542c0 : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":25},{"id":"f6a175d17ae2e1ef10c11eb4cfa2604ec408dccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-17 02:12:38.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"c08d5ae723be98e4f12c19e035f6aab233932e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 03:20:40.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e929b55609b4cb492c87121b9d83cc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb272d4caf741f6bbe9af5be7470d49 : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/a5d20474e518498b8866c72e6f444ccf : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe4fd40123c84e8b8c3aa75892211099 : SUCCESS in 1h 01m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5261c13524c4ca7822d57688de43018 : SUCCESS in 36m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d574883885f40ffbb09336fddfaad80 : SUCCESS in 33m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b05fc604715a4d89889f1f69afdbc666 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22b51aa6e3ad436d8336a4953e5bea4d : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5faa89f68ad84b76adf3712b76bd924b : SUCCESS in 22m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ad847e3269e4495af3667a744bb7484 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1fd8bf442e794f6fb4062cd3da756c08 : SUCCESS in 29m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba1e06b05a04575802e0fb39c88d25f : SUCCESS in 19m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e18bb9d44a6b4e48b4f7ba84e6d7ec91 : SUCCESS in 28m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7c776799a8fd46b1834380768c6565b5 : SUCCESS in 12m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/35720c9f49e3427598fe130c5b537939 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5bc2d2e0e8664b4eb523af5e50475466 : SUCCESS in 8m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9016c19402344df3b648e628e59c212d : SUCCESS in 9m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53d5f44f62374360b8395f21633c1786 : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf467470a40e48919932db4329cebd46 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e91fa8b17ec4c7a8810f62b98f964c3 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db800dcc142a40569af4c44bf3dbeac5 : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dccafcbe184d426689a54679e9fb7dab : SUCCESS in 44m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/306a1f5d03e9413dbfaa9b40f0dfdbbc : SUCCESS in 25m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03d113f8bb1f43b999bb76eacca18be7 : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da188327ff84ac9aa35b974af564715 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/443f23f8b0344b23a3cceb0f649ae09b : SUCCESS in 19m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a8bc58bd6e1407cb8e426a1c5ccbf95 : SUCCESS in 54m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/482c1fa25c6a43aba4721ce77b676f71 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":26},{"id":"a22e2f56995f301a36ec7bc242e5a53be4a335fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 04:40:34.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf45b693be34367b0a9ac0c73b9c131\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2db38e289f79487f98dc682cc1d31d50 : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":26},{"id":"444e88a80a23e1c92dbd617a64be343641e8e0bf","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-08-08 15:31:22.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"e056fe6b5ab06f9dca2da0778c86510b3e314b3a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-08-22 01:41:53.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"6e269fb7908f1ef903e86e5cbbea4fb1ee2d0ef2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-08-25 05:27:56.000000000","message":"Patch Set 26: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":26},{"id":"296598e1afab507e655abc33fb68e07d2c710ef0","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 20:23:13.000000000","message":"Patch Set 26:\n\n(7 comments)","accounts_in_message":[],"_revision_number":26},{"id":"4c6bcbd1e1390b989befe9ef88880e3442b2836a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 20:24:47.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"31e57ac00c553c9f633f96c91a1b402bc06a3d8b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 21:01:41.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"9037a9cfa25831cd04737c4683318a456bad597a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 21:36:21.000000000","message":"Patch Set 27: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c4c4339334e46f89d7aeee5213e6d75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8af4dbaa6244e0fbf55ede06a6921f5 : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/452bd5a59bf24c868b5337eb0c79b8f7 : SUCCESS in 1h 06m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f46b2eda04b84d76981a0cdb0bcee5f2 : SUCCESS in 49m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6ed07f17016429c846676ebdc501248 : SUCCESS in 1h 03m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40dd765360dc4b13903442edcf7f6127 : SUCCESS in 33m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/607b35b4707e40cbbd0f9a3f733f7373 : FAILURE in 6m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0daf407356514f0b88c7a48d38dd88e1 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec37384f89a545199902c409c7d635c0 : FAILURE in 4m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c43b96dee1d348efa8c16a75539d3598 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/065fa64a3bc143cd8641c4e2724b1047 : SUCCESS in 22m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06ef03a8c03841b39824ea988436de51 : FAILURE in 5m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/125785ff778a414d96250d0e77d73994 : SUCCESS in 22m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/48df8d854f0b4464864dc12140d0a479 : SUCCESS in 28m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51b07fd1d8e0446c8dc730cbb7036670 : SUCCESS in 1h 05m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2023658edfe241bebec20ce33e13d165 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fac2eeede4a3406985a6670918be5e9f : SUCCESS in 8m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05d95fbfb3664f14830682c09a0d59e6 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5dcedaad76a04546bf7f12006a459961 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d9c459bab69482a9588256de15c239f : SUCCESS in 11m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c679bfb5deb74890b0d6125201446d9b : SUCCESS in 56m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e91537932eb34726945ff748ccfd4046 : SUCCESS in 39m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bec462146e7402c9311d1cb805137f8 : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b06198647464eab92fcfd1bb2ad6cca : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8431fb6b85d6408292c6c373191e38ce : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2343f008007942f7bceccfd4e456e5bc : SUCCESS in 15m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae628ccff9a046c4ae0067fb9272f911 : SUCCESS in 47m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1244b2ed20fa438bad5368cf10f4b293 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":27},{"id":"b703c96237747fde3ee4d7aeeedaaa86f817aeaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 00:40:25.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a487eb198548c0bd391d09c4af8eb4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dedb9f8bc8974f7a9f0a794b2d1e5bcb : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":27},{"id":"81b843ae30588c9a006a4f3ba3a2def25fe7180f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-02 13:20:35.000000000","message":"Patch Set 27: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"a57d4dbabad8133efe9be18344a18ea2fd9c42c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 19:18:54.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"d03a9021f43e535575e690216a1362e382eca34f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 19:57:35.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"48d7cf1953a1b53bee1557b7ddcd1de1fc15d85d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 19:57:59.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"f34c7aa4230ddea363c7fdf879aea5debc33abf9","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-02 20:38:48.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"76a662d33be1533f16be39beb3007b8a2eb4efbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:07:08.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120449efcda7481a9455d328193ab609\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f14542bba1e44aa9b432799a5e66b8f : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/da321512457f446fb03e17f7dca419dd : SUCCESS in 57m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebe909b285c943a6b9f0797f800d10f8 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0933292bf72f4a80806138c91bb71afe : SUCCESS in 32m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bea42ae3c954cfca9194660d1b79250 : SUCCESS in 1h 05m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/feb7555e1aff4cb88c8e6b09b7f1666c : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de856d765b37412fb08b09931cb7fc37 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edf54d03f5ad4494a161352cc4ac78b9 : SUCCESS in 23m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a18d7917968d4a32a7aba9cd4d9fde9c : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/90fc858184354feca74dae2d5e90f561 : SUCCESS in 19m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42eea33add17481b88a51941df93710d : SUCCESS in 15m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5af34ff57aab46a1a8226fedd82d1250 : SUCCESS in 30m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/30c197aadeae498da01a020ae40dfe29 : SUCCESS in 19m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8d3b6e6d85c4b0db2e8072eb932b2da : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5edaf1e8a7304f0b913f41e41e8a766d : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5a593ca672a74ba7b9bfad992e040c60 : SUCCESS in 12m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc3602e2dcf24b72ac81cdae8dadf46a : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b29350a3926845829643c1b9a3b1d14b : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb275517cc4848a090c83aa63208a37b : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e895ea30284ffeae8813f77297ba47 : SUCCESS in 56m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e76ca3096f04454b7b46beefdbfa5eb : SUCCESS in 56m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf41f717999e4b239a2fc62dfe5cb4b5 : SUCCESS in 23m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46cc5694d4a4402db1cd8a675b837a1a : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e119a6b43014ac7bdbade0f6524dc68 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d9c34b6c2bb4ae5a92d61bbf8ed767c : SUCCESS in 19m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0b08ca069ed4bdcb775c73c9a9b8560 : SUCCESS in 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/031d8211590249bdb944219e48ef136d : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":29},{"id":"b84c9fa4578e7d27a2b79637c07d34874543baab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 21:18:56.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aadbae21ba549c6adcdb9ac1adf9805\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d35e34f774349239027b01a6a81285a : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":29},{"id":"e01188f848420a14daccb580e3a47aad83a591bd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-03 14:44:41.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"960393d79a439e477fec933fa5b3a88fbf7cb2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:27:54.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"ca379e4680fdd139c962b1f3cbbeef0ec645643d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:28:12.000000000","message":"Patch Set 30:\n\n(5 comments)","accounts_in_message":[],"_revision_number":30},{"id":"1dbd9e76c5401a49edbae3a1582c44f6599484fa","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-09-03 19:28:53.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"dea5c2438110929553deb9a7da6df4ff3cf1b66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:38:31.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1739d4ef32642748cc616062cc7ecc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c613969b24b4d4f8523840e73483d42 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/31ea0d31324f415fb17be0c8a97dc35b : SUCCESS in 53m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36ef295134504e9c9310f7a8a328fa3c : SUCCESS in 56m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58fe2b6b6b854d6f969823903905f96d : SUCCESS in 47m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0039c257a8ca43ea9e758f2ba42766f5 : SUCCESS in 36m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20092c82e74440ff8b7cdcc84cd6c900 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adbc9e21b7d44be8b774405b92f0d36f : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc2493598d64edb911805fdb862822a : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a99cba6905f4c549fd96430297a23e9 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b927e0d6c9544c5b079cf704227eab6 : SUCCESS in 18m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65a945071b8f43a3985856516d43cbe7 : SUCCESS in 22m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6f2886ff84b4f8abf7fccab0d468fec : SUCCESS in 10m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b74d8d725bc64496aa7630a0271a4f93 : SUCCESS in 27m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ec0d1d78475c4bfcb48190a67fb0db55 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d518297eb744eb7944bd189c99e38c5 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df675e7627f146fcbf9e4cbc2a86dd92 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b449892f2a7452f951423e4da789da5 : SUCCESS in 3m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c7d4dab72db471ba752bc9ad247e734 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f11d551b89547508cab474bbda417b6 : SUCCESS in 10m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9185984a485f48c7be66827e513d0594 : SUCCESS in 51m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/572df691dd6c4b7cb096d3ceef952371 : SUCCESS in 58m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9a005405bd2406e8c8b84454a7a1670 : SUCCESS in 25m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41dcd1ca7663491c8be16aa2fe550c39 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efe4835068d4c659e1a99501280616e : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6af556b4e75b4b78b1f9d7d1b16e1618 : SUCCESS in 18m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b926da94f20548fea24c1095783117eb : SUCCESS in 37m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6b90afa03904f3ea152c685580212eb : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":30},{"id":"51395f890da7318fb3b838f443803f24d6f11bce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:00:59.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d83d74b2b8d41719eea1a379d7170bf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12d10ed8152147b08de8f8385269870f : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":30},{"id":"e545ff5b1d40fff9e02c4aebed764fc5e21a2bd0","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-09-08 19:34:05.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"2b64312fa59807dc5bf46a2720c5044de8ab46db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-24 17:32:57.000000000","message":"Patch Set 30: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":30},{"id":"415809a6a8963b5574861007e359ec80c271596b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-24 18:15:35.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"6a1623ca18b2dc4f85a3bdc0b0d1457b260fe0ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-01 14:37:31.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"87516de725a5465c6108aa4d7167af651ade7fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 15:42:52.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c26fd79d4554462a24a88c36ce6f763\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9050e0bc8658456d9bd35fe7491375bf : SUCCESS in 5m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/fcd95ebc50514bb7964d39e84501bc9a : SUCCESS in 1h 00m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/beac88a5994c4262a3c3ce8a84c18462 : SUCCESS in 59m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d319fb19abda49778e3d0ea0639a9ed9 : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/663fec09121544a1b0709e414c82de1d : SUCCESS in 32m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/355aac845000495f827e26fd1533a246 : SUCCESS in 22m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/087372730763478288020b86483ae121 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d61c430af88c4b50b9892d6bae5ad4b0 : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9e4a9b04af047fb90912e71f786b935 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/362fce90a401488ea55f9f91ac575790 : SUCCESS in 26m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f77307f33782470bab0f1c7c79f91d48 : SUCCESS in 16m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ac098b0970c4ab2acbbe73fb3fdd8c9 : SUCCESS in 21m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eb1aa901664b4d519e424d373edfc0a3 : SUCCESS in 11m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/478764034f0d4b23b87f07540b099c05 : SUCCESS in 7m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e1705f8523048c682d8c6d81aa79740 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e778a97a2a44e5c85b6ddb8e4912deb : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/addc2a155f934419972c39cad333f5c7 : SUCCESS in 8m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c5d3736bb4c485cb294abe2058b30ba : SUCCESS in 8m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/33156038fb564e9aa6f2ed979453bb38 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ae04cb54fd04b2baf0784630ffef867 : SUCCESS in 59m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d02fe63a3dca456f9385ed6b5270b1eb : SUCCESS in 54m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5661d840f95b4c9f806a6dd2d8e0e795 : SUCCESS in 55m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a5bba1dc25b4bf0bc5b6f1542f42cdc : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe516d188acd4b30ab05709e0c6dcbe0 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee3f05b46b93461aaa3f36bb7565eff8 : SUCCESS in 11m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b908f3bc9a0f4f6e8a69ea4c6e02cbbe : SUCCESS in 28m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d8ba232d3e3454384537d46ac4c9cb5 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":31},{"id":"2440c94a5abc473095bddf954a57bfe1a5775a69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:44:17.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3955b69f5748b1a6394cc78b3df511\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04bd66ccfe824862967bf8e25cbeecfa : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":31},{"id":"2d6fb5c0a24c1a924f523eb65e2b22d7f6e97d82","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-02 16:09:37.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ffc6b8d746e6df8c604ac38fcf259734f346a572","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-02 19:28:53.000000000","message":"Patch Set 31:\n\n(5 comments)","accounts_in_message":[],"_revision_number":31},{"id":"775907646d340761fe1ac22866f210d2069759b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-02 19:44:09.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"a73eac7da191f8d8d486784d31d6845bc5154a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 20:50:52.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb9dea42b2fb44c7b8601400244b6dca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e893076cbc134643b6ea317a3ed3bdcf : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b62d979f471a402ca8944025bb76c056 : SUCCESS in 55m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/290984a589f840d39c855885a2c9cd01 : SUCCESS in 53m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56fb818f04b74a3ca4bc684e60c07b71 : SUCCESS in 51m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/888120debc0641d580ee58692bc2a42b : SUCCESS in 1h 01m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db9a9c7b2a5b4369bc31dbcc4c644006 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b2beab74d78446282d4a1c8e6f29818 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c23b3cbb8e3541b39ba395878f2029ee : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b15eb7f3cfcd4a3d944850acfef054aa : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d6dfcac0ff4486cb675b1f5ce212677 : SUCCESS in 26m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82970a2368a043fca53fecca1a38cc0a : SUCCESS in 10m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/105326ac3551426c96c2c47df87b0579 : SUCCESS in 10m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c62c25e8fe84491686a27b4d66c3c4a9 : SUCCESS in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3dde70dbf6d84b6c89328e821bfb64e1 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90914c9b86ef41f5af4495b0dc22147e : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3bd38cc01894a6baddffeb2791aecdc : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33c86f81e1e34e069abb5b08adcb8c4b : SUCCESS in 4m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e4d3020126b4b62a5b499910211bd06 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd8e953e91244e8395a7e280d6ac11c5 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7e773f1f2f54e67b2d41452e518a335 : SUCCESS in 38m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f530a21a03d04f3894503a0a05213c4d : SUCCESS in 49m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/477df2afb1f14d55adf78b2b2ea7d0f6 : SUCCESS in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf36955737024508928af7772ed97572 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664b04dc3951469dac94955d3bfc1dda : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25115b0307d44f6cb32e858902a70568 : SUCCESS in 25m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f14b519261945ce89dc3cfa7728c720 : SUCCESS in 52m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5da9d3cb928841fcbf921456fc7bf188 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":32},{"id":"661e5bbfbb5df156d4ae67e7a0e511fda84eafcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:21:16.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9086379fc0bb4c31a8837189cb62f23a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7db1f955195549e4a5f1d45f90593523 : SUCCESS in 1h 15m 01s","accounts_in_message":[],"_revision_number":32},{"id":"fbc96bb2a800424d343bb564f1a0579640d8fd50","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-06 01:25:35.000000000","message":"Patch Set 32: Code-Review-1\n\n(3 comments)\n\nI really like this refactoring and cleaning up. I just think the commit message and import of the moved Exceptions needs to be fixed. Then I think this is good to go!","accounts_in_message":[],"_revision_number":32},{"id":"5ab8d9bfc418fe23659b5831598dcbee1e76c071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-07 17:43:45.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"3ffc957e82a5b5c27117488b20fcee07f599336f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 18:52:00.000000000","message":"Patch Set 33: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79012f5243f8423f8443c928723209ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6690215ddaec4248a4db47f1c4f72c44 : SUCCESS in 5m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/5688ff959832482eadf373de6c545c39 : SUCCESS in 35m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa10f4bcf1b249fcbde0ebf2ffc37040 : SUCCESS in 1h 05m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4a92a0c9fc2427b83f3d591a2027a8c : SUCCESS in 52m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0046ff176ee74d539f2956144b5f63bb : SUCCESS in 32m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30781815f7db49c7b5d4ccdd3ad71f56 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af6239140c884bb79e5170547bf671bf : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/777ca84330e2494891e5bc156213512b : SUCCESS in 21m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e16975f6c61441b9b32b18f426d5341 : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b554b535379f4746b0631117efa02dce : SUCCESS in 23m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d47776b3cda64f05b5063ced1f3aa629 : SUCCESS in 23m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/098f30b118e34eae8de6a3d74ab7f2d6 : SUCCESS in 27m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d61ca5fbf8694defa29ef43274bd25c4 : SUCCESS in 11m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5763c83db2164288818d2022743e2838 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b0c01c1749e9425c8c9044dc5674f408 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eabf1c280ccd47fc907b86a4c8eec4b9 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5ae84ef44264a5ca71d1a6f0bd51b7e : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78fd9caa335d4d259d238bcd8fc8074a : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ada2b845f6ed4c43b78e885bceeca1ac : SUCCESS in 4m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/89938b2f8c344ccfa4d55227c869eb62 : SUCCESS in 41m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/256a023c94ff4ff2a2290d84623aa8bb : SUCCESS in 21m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74d2530a7fed460d91deeee026cacd27 : SUCCESS in 22m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fde272d0b974051b18aa735c0302a7d : SUCCESS in 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9ae1a90b864764b1c726ca3c290b66 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db503c8851664840a4848ec2c3b24636 : SUCCESS in 8m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1977320acf545899ce371123998dbd0 : SUCCESS in 49m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79dded074dcf41af8600c43bb98ad2bd : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":33},{"id":"ccea45ae11861fedc9422f7ef96d94498c836b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:08:14.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861b4b00148c48ca91ca1ece2532328b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ded4db756ad04e3bb3c4a796e9ba43e1 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":33},{"id":"8818adbb2a3c8a45c84759fa8293ecccef910c4b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-08 18:45:02.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"75564bddea9c27739b6a1c2ed3bb997dbd4217f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-08 18:46:11.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"2a945c74e3b44122f9fec9e9c7f650f763a2f122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 20:00:04.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94ec8732c0544c15b32b85720c7759a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b4859aed72489e93db3fe4ff277393 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/3a45d0cd8aa948ce9a5801e7e3644435 : SUCCESS in 29m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df74257439d64053bd18383b9ec67544 : SUCCESS in 1h 08m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee65c15b45504774916f129fe653addc : SUCCESS in 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/745208a0866c425ea3ece7ba5d4fc715 : SUCCESS in 35m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5c72d086a044085802a1d6e3c948068 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/237a5b18f16745b190e87ffda1468eba : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88b6ebc3d0484575865110960f309f1f : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f681dbec5f45499d81e3fd1e3f604017 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f36b1f1fcdf04be8893a8d32beb107ab : SUCCESS in 25m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2a21eb52e80488fa2146c82b8e7d68c : SUCCESS in 20m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e7afd0066e4e4aa788d31b1af4078e20 : SUCCESS in 28m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1a6aabf01b024c30a1edf63bbd22ae14 : SUCCESS in 24m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7814dcf9161143f4a6fdd0ab2261cb46 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/15387437b8024877a5397760b4b49baa : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7dfb8b98a5d242ebb505f56358076528 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/969f274187c24e9bae5fcc61197f949e : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bb9bddabdee4259a213ccbd9800de8a : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/05a6b94fa8c643d89be30ae52cb8ca86 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4cb03ea6e844c9d9a04bc9af9c3f6d8 : SUCCESS in 1h 05m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f949fa41d0364edf83caceff9dd99e83 : SUCCESS in 48m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac3546940c164558adb6381fa2fbdc75 : SUCCESS in 1h 01m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06bc81924fbd40c2ac33bbb1555c78ab : SUCCESS in 27m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1dcaed64e9441c8e60fa303ca84112 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dbf72f5f258405cb4cab83cc3c371f7 : SUCCESS in 20m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28b031a192f543d495b307fec1f06cd1 : SUCCESS in 46m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b8def37a9c14bf39965a7d0dd9f5cf8 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":34},{"id":"147b4925f987a36efcbafd460d84059414c5ea2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:14:33.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65aac42d850245558218d4db47ea8b93\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c49ae2d0eb5483fa00e02177484cfdb : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":34},{"id":"ae40eb4e74734ff61ca67b2d413f55256af90e29","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-09 10:37:58.000000000","message":"Patch Set 34:\n\n(6 comments)","accounts_in_message":[],"_revision_number":34},{"id":"4191641e54ddcc0c20a39c2668d5690f2d89a811","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-10 16:32:35.000000000","message":"Patch Set 34:\n\n(5 comments)","accounts_in_message":[],"_revision_number":34},{"id":"a7fe3f71e19beea9e3f111759e495b3271ba5494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-10 18:26:16.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"c04ba38ec63e7ba5a9e4a8267ee121375acf0e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:16:50.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a05dbca010b46578a5c6f694bab243a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b36785b932004516ae1e2fda99ee341f : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":35},{"id":"233c231d233db5bd5c5387274803bf865a5cbc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 19:36:44.000000000","message":"Patch Set 35: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6da967a0a84b4a2fb12d270e5299c501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7540b92b50447c8d3466f1aaf0ff88 : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/91a083be29e14656a45d2b6229707a6b : SUCCESS in 1h 00m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/926cd99339d44a65a85f8fbffa520845 : SUCCESS in 32m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/03ff03228290431d90bd9103e79d3095 : SUCCESS in 53m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12df0f68ae314f11881a15e01a3f9d48 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16bd265725784c99ad131e8430f4f8c4 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cebd60a231024a13a3692cfd42036949 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e58720572594c8fb8b20c5d2f4944dc : SUCCESS in 10m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5284f2671877421eba8665cef9d6495a : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/680a855dc51e43df869311e9f6dec845 : SUCCESS in 22m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4f1067711d842febc002e2a4295d650 : SUCCESS in 10m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d245a6a4ca8447289d5af5abdca2ce8 : SUCCESS in 31m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d19bc95f54d4e018944877f4ff2c28a : SUCCESS in 29m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c41ffc851ca14da8996c7999a1d07f98 : SUCCESS in 7m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/afb558d6ccc94666b05ad47ae4a8721c : SUCCESS in 7m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3449d1cbbb1a418cb445540487bd6126 : SUCCESS in 13m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae38dc717e9b4fb0a715d855f3ad4a48 : SUCCESS in 4m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2d4babc08c741388bffa35f4b208264 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a94feee443084cceb79a6e9e04dc2b18 : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a6aefb8e5e345a6b2c7aebd73fae5b1 : SUCCESS in 57m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d92c7b7253fc4ba593336b052d7c39e0 : SUCCESS in 53m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49828b5bfc56408a82c877ca3fa01351 : SUCCESS in 34m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4427bed74bc94c64bdff2684410ee398 : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36927dc631ca4c6bb6a30e7deecd3ce3 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8ea6a6e7e7d4fc19e414461d6570872 : SUCCESS in 19m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be4001c251fe48a4ac354ef1bd80e991 : SUCCESS in 53m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fda1960ff664ae8b58439a842213469 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":35},{"id":"826a479294984c530623fd1470041f7c78860a8a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 12:27:22.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"e9e940ece82532869b83856df5b837d0047cfb4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 14:24:15.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"0eacfefc4eb13dbd31c40914953740fe0bda6ada","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 14:38:23.000000000","message":"Patch Set 36:\n\n(6 comments)","accounts_in_message":[],"_revision_number":36},{"id":"1e48327ebc53f3def1e310bd00b4bac62cb9ac0f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 14:55:36.000000000","message":"Patch Set 36:\n\n(5 comments)","accounts_in_message":[],"_revision_number":36},{"id":"30403703a1c2cbb9abc989e70720ea31c3ea5f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-13 15:25:49.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"92aac2090cf23575a4c55a656f020520a61fbf6c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-13 15:33:59.000000000","message":"Patch Set 37: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"a55f720ace2b2e0c9d1cfcce0bcfb4a0c0e49c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 15:58:48.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab3b572dc584276a3372df3432dda98\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fc67da348214c93b33ba74121b6b070 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":37},{"id":"e2aa47309c5c9f98a548c5bfc70055db924064a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 16:33:28.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cee8f82d745a4ac693b2201c530301c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a444f0a487c4abe811d4bb27a052d46 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/ba0ed9fa9bee4dbcb73af28cc6357ad6 : SUCCESS in 34m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c1069022e7b495cafea292f23d157d1 : SUCCESS in 1h 02m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2dd9c54bf85d4ac4af9845d2eff429c9 : SUCCESS in 57m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba1b64759e134149be2c2bbc2708a01c : SUCCESS in 31m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60e3bc5ae31545529441b87646e710f5 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/157b5b02a19742b89944ba76ee726425 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95147620928e49d18761ac4082f9361a : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24c318b2acc84597b7991deeff00b511 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5443786dc1704391908ce656db2b242a : SUCCESS in 25m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39c2308413524b98aea085337b4d2cc8 : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c6e4c46efd7f43a09980f8cbc07ecd7b : SUCCESS in 29m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e3b789307c04e23809eccc662fe602c : SUCCESS in 13m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bf4f86350b3344d48dfa137d0c656fe8 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cf1dc32769894820b2834557e206b92f : SUCCESS in 13m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8002356a16f84f13a9fd72162f1954e8 : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e836fbfb99b4844b540cac56b9779b8 : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f79e7fea96214b1d95330b804f49bd30 : SUCCESS in 13m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1aa86b9de65f4fcea3ff4286c0176375 : SUCCESS in 9m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3de603d363e34be980147131e98621dc : SUCCESS in 57m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48e047410bfc4c8ca4916b84aadeeb67 : SUCCESS in 26m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2dc2ef95b814fee93fb812561ed7236 : SUCCESS in 37m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a8b2f005bec45499464edaa2387e9e2 : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314cb334461c4e6c8e05bfbf6a3ef2db : SUCCESS in 3m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ff02f852aa54aa38838f38750233de0 : SUCCESS in 18m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6989cbf7e084c30aabd3035f6e25899 : SUCCESS in 47m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d91499a601a6484489fb9611d66095e9 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":37},{"id":"ecf2c3227b3f96d94443b8c5c3d9338606beea55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-14 13:58:00.000000000","message":"Patch Set 38: Patch Set 37 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"90d8aa7d7d9e229cc8f0255dc8cffc1821afe3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:15:12.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8225c13f9f9b4e88baf7ea37adeb55ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a764cd0a1ba4471cb60e3527d11d18bd : SUCCESS in 6m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/181ce361060142ec8234a2d7b7005973 : SUCCESS in 26m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df05059fe10c411095d0bc70b7c3641c : SUCCESS in 1h 05m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/972f725735bb40929109fb7f75f354bf : SUCCESS in 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1cfa2e1525b494890c50ab9d8c314e5 : SUCCESS in 42m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/008389ab3c7d4ef79afd9867af6f4248 : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/454ea5e51e524fb2a4025773cd1b12ea : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/590a48d2998e4a90b0d5914185fa98bd : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/334f3d9373674fd2ba1c02b45781f1f1 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f0f75d99e5f4d8eab518b008873b2a3 : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba0b2979225d45e598ad413665ea2573 : SUCCESS in 24m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/477cb5a6a4e44b0395b07b7cfe997110 : SUCCESS in 20m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2616b61dccd14c5dad56cdfcf6339f03 : SUCCESS in 34m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/16b7f3fe0b0d44a3b6c5556cbf24572d : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/050b618098b1404c967f98c63b2d529b : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f167fe6f49de4e7ea4716627c87604dc : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d60630de91f141e0af2302f438f79cb9 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/333345f1740f45ee8c5eede936a13cec : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c581c8298e64069b9e668a720a18196 : SUCCESS in 7m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e89ec56ec55b4e50929e9780fa8063fd : SUCCESS in 56m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3400beaaf064f828704c7b3a98607ca : SUCCESS in 21m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fceb38c7b674bf8b8537121cd961e96 : SUCCESS in 32m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0ed07fc5758456a94e1c75bb1388c02 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5593d80cd2944414ad348f4ecc1a04e1 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5deb4676a50c4640bd2bdeb273b12f70 : SUCCESS in 18m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a9992d33eaf4d0ca619da71031c4752 : SUCCESS in 42m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa171c9743b541409c49f2446dd4664b : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":38},{"id":"7d22d618f8c7db72e7053c2e7134a213c6876135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 16:08:08.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c53d9b159f4af189ec57a18789a8de\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02d7d56470ac41558bcbf6a944fefbb8 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":38},{"id":"8106669255d0b5a30ee795ed23d30775d01369c8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-03 04:31:19.000000000","message":"Patch Set 38: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":38},{"id":"aa7481d812ace92016578c247c0417d36026d018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 04:32:16.000000000","message":"Patch Set 38: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":38},{"id":"5334aca4fa3389f601330e79be115035731ac565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:37:39.000000000","message":"Patch Set 38: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9dceea81ec4c3ba3e055b8d7f66d33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7a8677553545f8be7f5b5589719d94 : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/455c8db8b15448e4a7306eb203ca30f8 : SUCCESS in 56m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cc60c44bf4a4b919c45d6f5517ffa4b : SUCCESS in 43m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc4f40a9cbd44f0a80a4c2e008c0f206 : SUCCESS in 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6abf423cab93493795eed7fc70017c9b : SUCCESS in 1h 01m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b103eec288bf427ba8e3a3e4dbc41cd1 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47147cea4ed6480dbe433eacdcbd4b08 : SUCCESS in 26m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be8ca88b654d4b9790f94c140e4e3cde : SUCCESS in 9m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9fef372e7da44cb8a3ba43ef5216da34 : SUCCESS in 12m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/83a2e3fac91a4c2aaddd886f6297eda0 : SUCCESS in 30m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ecf4e5d90a684ec19eb927cb51970cd2 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba30f86d0a044036a0ccdc80260de930 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e9e1be601804004b132c21b905a09c1 : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb34397c094d43d8a6aafeab0461ce00 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aef658e6dd7246799f9f094dcaaff6b2 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a43a778d91b4e7abeac31c50c2225a7 : SUCCESS in 35m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9c9bf4960b749fe8a2a9be68c823ae6 : SUCCESS in 49m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b147c38dce9c4d52a2c0389e8f80cc76 : SUCCESS in 24m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db7f4242e6dd408b983fcd9fd6fe9e28 : SUCCESS in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aed0881315446928269653c64d9fbba : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62cb2a46a7d546d2a15c1d0b724aea09 : SUCCESS in 18m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9090be7b7da4a3c915b3b56444b9e46 : SUCCESS in 37m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f19a2da2ea814510a7fd6633577e0ce3 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":38},{"id":"cae6c6e6e9972f8fd396236816d0de2aa9a3b4e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:37:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":38},{"id":"e7624020c9d106844b69bcc91358465bea99470f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 05:38:33.000000000","message":"Patch Set 38:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f90b76c18474dfb990bb9c79a3b83ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88246492e9264c60a1eaac494e9d28f2 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ade916d029d94cf99fc62654a631a992 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":38}],"current_revision_number":38,"current_revision":"9d7e7e27a5114ee7390519324514e1bd26124ffc","revisions":{"50db954da1b00a259211cfb4e1e0e2d0e3c9fda1":{"kind":"REWORK","_number":1,"created":"2025-02-05 15:00:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/1"}}},"commit":{"parents":[{"commit":"714cd71ca4665962e63eacd1242a6281ff6f1851","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-03 21:35:07.000000000","tz":-300},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50db954da1b00a259211cfb4e1e0e2d0e3c9fda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50db954da1b00a259211cfb4e1e0e2d0e3c9fda1"}]},"branch":"refs/heads/master"},"0540a79255c98653ba1f53aee8294cacaf33d4e5":{"kind":"REWORK","_number":2,"created":"2025-02-10 15:03:00.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/2"}}},"commit":{"parents":[{"commit":"714cd71ca4665962e63eacd1242a6281ff6f1851","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-10 14:39:46.000000000","tz":-300},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0540a79255c98653ba1f53aee8294cacaf33d4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0540a79255c98653ba1f53aee8294cacaf33d4e5"}]},"branch":"refs/heads/master"},"dc2812e2aab06c8082ae921261941360f7a7cc6c":{"kind":"REWORK","_number":3,"created":"2025-02-13 20:59:10.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/3"}}},"commit":{"parents":[{"commit":"714cd71ca4665962e63eacd1242a6281ff6f1851","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-13 20:21:43.000000000","tz":-300},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc2812e2aab06c8082ae921261941360f7a7cc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc2812e2aab06c8082ae921261941360f7a7cc6c"}]},"branch":"refs/heads/master"},"3ae4fb5a88f17c5bcd5a6a3208e8fd990c15198d":{"kind":"REWORK","_number":4,"created":"2025-02-27 15:00:48.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/4"}}},"commit":{"parents":[{"commit":"714cd71ca4665962e63eacd1242a6281ff6f1851","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-02-27 14:29:48.000000000","tz":-300},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ae4fb5a88f17c5bcd5a6a3208e8fd990c15198d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ae4fb5a88f17c5bcd5a6a3208e8fd990c15198d"}]},"branch":"refs/heads/master"},"af0326340dd917f38277eb23a1f576d6e8cbd3bc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-05 14:58:33.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/5"}}},"commit":{"parents":[{"commit":"6f89f8f9e279e8ae7b9b9351cbd5c529d9128ab1","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f89f8f9e279e8ae7b9b9351cbd5c529d9128ab1"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-03 21:36:48.000000000","tz":-300},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0326340dd917f38277eb23a1f576d6e8cbd3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0326340dd917f38277eb23a1f576d6e8cbd3bc"}]},"branch":"refs/heads/master"},"f4fa65eb9bd2da286313aa0a1f1840f15d1fcf69":{"kind":"REWORK","_number":6,"created":"2025-03-07 11:58:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/91/940791/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/6"}}},"commit":{"parents":[{"commit":"eb097be036baedb4e4fa9e8bdb42eb470bc563ee","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb097be036baedb4e4fa9e8bdb42eb470bc563ee"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 11:55:33.000000000","tz":0},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fa65eb9bd2da286313aa0a1f1840f15d1fcf69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fa65eb9bd2da286313aa0a1f1840f15d1fcf69"}]},"branch":"refs/heads/master"},"a215fefd5502c2ce78eec66b502678cdd64f9cad":{"kind":"REWORK","_number":7,"created":"2025-03-07 13:12:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/91/940791/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/7"}}},"commit":{"parents":[{"commit":"2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:52:51.000000000","tz":0},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a215fefd5502c2ce78eec66b502678cdd64f9cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a215fefd5502c2ce78eec66b502678cdd64f9cad"}]},"branch":"refs/heads/master"},"c2e806bf13053be39a9a5be1f1fdc000d950c80e":{"kind":"REWORK","_number":8,"created":"2025-03-19 19:58:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/91/940791/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/8"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-19 19:23:57.000000000","tz":-240},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2e806bf13053be39a9a5be1f1fdc000d950c80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2e806bf13053be39a9a5be1f1fdc000d950c80e"}]},"branch":"refs/heads/master"},"0d4e6fb793d99ee43ea1267c3087861b84c016c1":{"kind":"REWORK","_number":9,"created":"2025-04-17 17:27:10.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/9"}}},"commit":{"parents":[{"commit":"7788ea0d82ea4a8b34b6bab47756468733528907","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7788ea0d82ea4a8b34b6bab47756468733528907"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-17 17:27:10.000000000","tz":0},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d4e6fb793d99ee43ea1267c3087861b84c016c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d4e6fb793d99ee43ea1267c3087861b84c016c1"}]},"branch":"refs/heads/master","description":"Rebase"},"c3c7eeec4a0c0dc7f4ed6b0e03942e49cfbebefc":{"kind":"REWORK","_number":10,"created":"2025-04-17 17:53:39.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/10"}}},"commit":{"parents":[{"commit":"7788ea0d82ea4a8b34b6bab47756468733528907","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7788ea0d82ea4a8b34b6bab47756468733528907"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-17 17:52:41.000000000","tz":-240},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c7eeec4a0c0dc7f4ed6b0e03942e49cfbebefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c7eeec4a0c0dc7f4ed6b0e03942e49cfbebefc"}]},"branch":"refs/heads/master"},"432d31e9d30307005c25e85cd7845984d7b991f3":{"kind":"REWORK","_number":11,"created":"2025-04-28 01:18:35.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/11"}}},"commit":{"parents":[{"commit":"870c7f0899d7b76744b15096031e53264e161744","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/870c7f0899d7b76744b15096031e53264e161744"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-28 01:13:36.000000000","tz":-240},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/432d31e9d30307005c25e85cd7845984d7b991f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/432d31e9d30307005c25e85cd7845984d7b991f3"}]},"branch":"refs/heads/master"},"0c9c04902043ff2f0f0a57b99dcc81bc3b00de65":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-04-29 13:17:04.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/12"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-29 13:17:04.000000000","tz":0},"subject":"stats: Add path_parts environ for s3api","message":"stats: Add path_parts environ for s3api\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9c04902043ff2f0f0a57b99dcc81bc3b00de65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9c04902043ff2f0f0a57b99dcc81bc3b00de65"}]},"branch":"refs/heads/master","description":"Rebase"},"a8a64a3244f7b179a72d4e083e6ca2d4580575fb":{"kind":"REWORK","_number":13,"created":"2025-04-30 15:18:43.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/13"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-30 15:17:50.000000000","tz":-240},"subject":"stats: s3 helper functions","message":"stats: s3 helper functions\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8a64a3244f7b179a72d4e083e6ca2d4580575fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8a64a3244f7b179a72d4e083e6ca2d4580575fb"}]},"branch":"refs/heads/master"},"f5d205ac2ce0b30beaba243c5a866446cb202790":{"kind":"REWORK","_number":14,"created":"2025-04-30 17:48:08.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/14"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-30 17:47:42.000000000","tz":-240},"subject":"stats: s3 helper functions","message":"stats: s3 helper functions\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5d205ac2ce0b30beaba243c5a866446cb202790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5d205ac2ce0b30beaba243c5a866446cb202790"}]},"branch":"refs/heads/master"},"e300cfbdc8ffd07edac65c9aac2f94bc2ff16801":{"kind":"REWORK","_number":15,"created":"2025-05-05 12:20:12.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/15"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-05 12:19:48.000000000","tz":-240},"subject":"stats: s3 helper functions","message":"stats: s3 helper functions\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e300cfbdc8ffd07edac65c9aac2f94bc2ff16801"}]},"branch":"refs/heads/master"},"5885da7c26456ecbb4015384085ae84822b7be57":{"kind":"REWORK","_number":16,"created":"2025-05-06 18:57:46.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/16"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 18:55:24.000000000","tz":-240},"subject":"stats: s3 helper functions","message":"stats: s3 helper functions\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5885da7c26456ecbb4015384085ae84822b7be57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5885da7c26456ecbb4015384085ae84822b7be57"}]},"branch":"refs/heads/master"},"70bd16d5bb983854951e1dba4badc8738f2f0d6c":{"kind":"REWORK","_number":17,"created":"2025-05-06 20:54:19.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/17"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-06 20:52:44.000000000","tz":-240},"subject":"stats: s3 helper functions","message":"stats: s3 helper functions\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70bd16d5bb983854951e1dba4badc8738f2f0d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70bd16d5bb983854951e1dba4badc8738f2f0d6c"}]},"branch":"refs/heads/master"},"228dd4fffd6c88aa7a3b82e44af78866c0dd3fbf":{"kind":"REWORK","_number":18,"created":"2025-05-08 19:56:14.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/18"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-08 19:55:13.000000000","tz":-240},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/228dd4fffd6c88aa7a3b82e44af78866c0dd3fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/228dd4fffd6c88aa7a3b82e44af78866c0dd3fbf"}]},"branch":"refs/heads/master"},"5c4c62660acfd671c38966d9301ca78830b26f45":{"kind":"REWORK","_number":19,"created":"2025-05-13 14:20:50.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/19"}}},"commit":{"parents":[{"commit":"a5160909acb1339923039fd10ca4cf0e95253b4c","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-13 14:20:25.000000000","tz":-240},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c4c62660acfd671c38966d9301ca78830b26f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c4c62660acfd671c38966d9301ca78830b26f45"}]},"branch":"refs/heads/master"},"46b31ab3172a8c4c1a28545172157f4f46a38023":{"kind":"REWORK","_number":20,"created":"2025-05-15 03:01:48.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/20"}}},"commit":{"parents":[{"commit":"ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-15 03:01:21.000000000","tz":-240},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b31ab3172a8c4c1a28545172157f4f46a38023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b31ab3172a8c4c1a28545172157f4f46a38023"}]},"branch":"refs/heads/master"},"e21838a53a7aebf5ab16a33bd1b21bfe877b3d27":{"kind":"REWORK","_number":21,"created":"2025-05-20 12:39:32.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/21"}}},"commit":{"parents":[{"commit":"ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-20 12:39:04.000000000","tz":-240},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21838a53a7aebf5ab16a33bd1b21bfe877b3d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21838a53a7aebf5ab16a33bd1b21bfe877b3d27"}]},"branch":"refs/heads/master"},"f5546e62ad2ca87422c4898d0f556e43a38f129a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-05-21 20:27:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/22"}}},"commit":{"parents":[{"commit":"1cd20f87dec469dc4f7c9ac27d99ffec0b859301","subject":"Merge \"Add labeled metrics to proxy-logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cd20f87dec469dc4f7c9ac27d99ffec0b859301"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-21 20:27:52.000000000","tz":0},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5546e62ad2ca87422c4898d0f556e43a38f129a"}]},"branch":"refs/heads/master","description":"Rebase"},"5ef3533ba64da5510a249e464dce2f9bfc19172e":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2025-07-09 19:40:42.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/23"}}},"commit":{"parents":[{"commit":"1cd20f87dec469dc4f7c9ac27d99ffec0b859301","subject":"Merge \"Add labeled metrics to proxy-logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cd20f87dec469dc4f7c9ac27d99ffec0b859301"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-09 19:40:42.000000000","tz":0},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ef3533ba64da5510a249e464dce2f9bfc19172e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ef3533ba64da5510a249e464dce2f9bfc19172e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"87fc1c81272b6b287df8c022443780aa9c6ab19d":{"kind":"REWORK","_number":24,"created":"2025-07-16 15:48:46.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/24"}}},"commit":{"parents":[{"commit":"2fc9209d47fc33067a7f15c4b90c0fdbb5529484","subject":"Merge \"s3api: Validate additional checksums on upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc9209d47fc33067a7f15c4b90c0fdbb5529484"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-16 15:48:46.000000000","tz":0},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87fc1c81272b6b287df8c022443780aa9c6ab19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87fc1c81272b6b287df8c022443780aa9c6ab19d"}]},"branch":"refs/heads/master","description":"Rebase"},"237178f76cd2785c88b614edacf371db8e654c09":{"kind":"REWORK","_number":25,"created":"2025-07-16 20:46:50.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/25"}}},"commit":{"parents":[{"commit":"2fc9209d47fc33067a7f15c4b90c0fdbb5529484","subject":"Merge \"s3api: Validate additional checksums on upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc9209d47fc33067a7f15c4b90c0fdbb5529484"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-16 20:46:36.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/237178f76cd2785c88b614edacf371db8e654c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/237178f76cd2785c88b614edacf371db8e654c09"}]},"branch":"refs/heads/master"},"f7c69412ad30693b81983366b2d173f85296d6c7":{"kind":"REWORK","_number":26,"created":"2025-07-17 02:12:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/26"}}},"commit":{"parents":[{"commit":"2fc9209d47fc33067a7f15c4b90c0fdbb5529484","subject":"Merge \"s3api: Validate additional checksums on upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc9209d47fc33067a7f15c4b90c0fdbb5529484"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-17 02:12:13.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7c69412ad30693b81983366b2d173f85296d6c7"}]},"branch":"refs/heads/master"},"120e31e182803b97cbc2908e9786297ca5775770":{"kind":"REWORK","_number":27,"created":"2025-08-27 20:24:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/27"}}},"commit":{"parents":[{"commit":"2fc9209d47fc33067a7f15c4b90c0fdbb5529484","subject":"Merge \"s3api: Validate additional checksums on upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc9209d47fc33067a7f15c4b90c0fdbb5529484"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-27 20:24:28.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/120e31e182803b97cbc2908e9786297ca5775770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/120e31e182803b97cbc2908e9786297ca5775770"}]},"branch":"refs/heads/master"},"dc25a4c63297a9a5fbc9392364e63e3f7c147762":{"kind":"REWORK","_number":28,"created":"2025-09-02 19:18:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/28"}}},"commit":{"parents":[{"commit":"2fc9209d47fc33067a7f15c4b90c0fdbb5529484","subject":"Merge \"s3api: Validate additional checksums on upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fc9209d47fc33067a7f15c4b90c0fdbb5529484"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-02 19:18:21.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc25a4c63297a9a5fbc9392364e63e3f7c147762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc25a4c63297a9a5fbc9392364e63e3f7c147762"}]},"branch":"refs/heads/master"},"9b57c987181266a52a7a420694592b07856d4884":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2025-09-02 19:57:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/29"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-02 19:57:59.000000000","tz":0},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b57c987181266a52a7a420694592b07856d4884"}]},"branch":"refs/heads/master","description":"Rebase"},"d4a254ddbfade515959995043f1a577d5e8f8715":{"kind":"REWORK","_number":30,"created":"2025-09-03 19:27:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/30"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-09-03 19:27:35.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4a254ddbfade515959995043f1a577d5e8f8715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4a254ddbfade515959995043f1a577d5e8f8715"}]},"branch":"refs/heads/master"},"71942b83c6f3d96369aab9f2611b9a74aae8d1fe":{"kind":"REWORK","_number":31,"created":"2025-10-01 14:37:31.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/31"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-01 14:27:45.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71942b83c6f3d96369aab9f2611b9a74aae8d1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71942b83c6f3d96369aab9f2611b9a74aae8d1fe"}]},"branch":"refs/heads/master"},"615393f49ea53cccbc36321399a5d85caeca16d1":{"kind":"REWORK","_number":32,"created":"2025-10-02 19:44:09.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/32"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:43:39.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/615393f49ea53cccbc36321399a5d85caeca16d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/615393f49ea53cccbc36321399a5d85caeca16d1"}]},"branch":"refs/heads/master"},"656c5709c8025732ef3b84e24b17d7f24d36257f":{"kind":"REWORK","_number":33,"created":"2025-10-07 17:43:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/33"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-07 17:39:22.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/656c5709c8025732ef3b84e24b17d7f24d36257f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/656c5709c8025732ef3b84e24b17d7f24d36257f"}]},"branch":"refs/heads/master"},"17111ccac98839d9eb8a0144ca3cd97c6a418f2a":{"kind":"REWORK","_number":34,"created":"2025-10-08 18:46:11.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/34"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-08 18:45:53.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17111ccac98839d9eb8a0144ca3cd97c6a418f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17111ccac98839d9eb8a0144ca3cd97c6a418f2a"}]},"branch":"refs/heads/master"},"ddccb990f55e264ebfda4442346668a99302df5e":{"kind":"REWORK","_number":35,"created":"2025-10-10 18:26:16.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/35"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-10 18:26:01.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddccb990f55e264ebfda4442346668a99302df5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddccb990f55e264ebfda4442346668a99302df5e"}]},"branch":"refs/heads/master"},"5b1052c7bbab23750a4ab487e9d4ad5ee2869ad0":{"kind":"REWORK","_number":36,"created":"2025-10-13 14:24:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/36"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-13 14:23:56.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b1052c7bbab23750a4ab487e9d4ad5ee2869ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b1052c7bbab23750a4ab487e9d4ad5ee2869ad0"}]},"branch":"refs/heads/master"},"a3f8b0ab4573bf1588744d568fc49bc08c40f5df":{"kind":"REWORK","_number":37,"created":"2025-10-13 15:25:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/37"}}},"commit":{"parents":[{"commit":"21325988dfe7d63d1f3ef78b9af8a2297f787e1f","subject":"CI: Remove a bunch of unnecessary bindep profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21325988dfe7d63d1f3ef78b9af8a2297f787e1f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-13 15:15:27.000000000","tz":-300},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f8b0ab4573bf1588744d568fc49bc08c40f5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3f8b0ab4573bf1588744d568fc49bc08c40f5df"}]},"branch":"refs/heads/master"},"9d7e7e27a5114ee7390519324514e1bd26124ffc":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2025-10-14 13:58:00.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/91/940791/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/940791/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/940791/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/940791/38"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-30 20:55:44.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-14 13:58:00.000000000","tz":0},"subject":"Provide some s3 helper methods for other middlewares to use.","message":"Provide some s3 helper methods for other middlewares to use.\n\nget_s3_access_key_id returns the S3 access_key_id user for the request\nis_s3_req checks whether a request looks like it ought to be an S3 request\nparse_path returns a wsgi string\nextract_bucket_and_key extracts bucket and object from the request\u0027s PATH_INFO\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: Iaf86a07238cca6700dee736f55d4c0672cccf1b1\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d7e7e27a5114ee7390519324514e1bd26124ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d7e7e27a5114ee7390519324514e1bd26124ffc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
