)]}'
{"id":"openstack%2Fswift~940840","triplet_id":"openstack%2Fswift~master~If29c420c02f375152e6292c67d0d9b49a4a8e677","project":"openstack/swift","branch":"master","topic":"ic-fix","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-06 23:17:37.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-06 15:46:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If29c420c02f375152e6292c67d0d9b49a4a8e677","subject":"WIP: Normalize HTTP headers consistently","status":"NEW","created":"2025-02-05 22:13:29.000000000","updated":"2025-03-03 20:02:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"c63f3a73b73e7395073d6ed5c6d6b2cccc7a48e4","_number":940840,"virtual_id_number":940840,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-06 23:17:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 23:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-06 01:32:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-03-03 20:02:42.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"bda4da401fd9b0224371f12f0ae7485844361ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-05 22:13:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38f6e1c64a8b7a3733c9d382ec9108233538b896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:27:21.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffd53d8a94af4a7ca4a0a93e3c7c9879\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1d14a442904affa8e8d7ebfabaac2c : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c94344eac5a747f3aaa32285ec609fcf : SUCCESS in 54m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5565d42347de4f3a856b332d03f2f966 : SUCCESS in 1h 07m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65252c93786748b8810058cf20ce1a37 : SUCCESS in 50m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d554bef4ebc3402480559acf7a154936 : SUCCESS in 1h 06m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a37fb743aa546869a283b8db3355ac7 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c6a4e6e8fd4411a8afc98405e0aac5a : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6febc9596efe47899ba6c0121fb4a733 : SUCCESS in 12m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e91e77e2f67b4d908fcbf71c6fcd5666 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b318a17b91454fe3af8be3160a61afd0 : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59788db28cc14bee9baddc2e75f6c931 : SUCCESS in 14m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a0446decd8f04b81b9e27fc71950950d : SUCCESS in 22m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bee46b11f8ce4d599b1f7304e04733b3 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b7c480bd29c4658944b84143e6ccaa4 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e0599e116694510bd2d5b7dd35b2d9a : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/513f09cc476d4baf8e386f2a89d7a662 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1cff4db78e94fb78c6e3ee3390a8895 : FAILURE in 15m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d7696b6e2454c5eb6b67eafbad8134c : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ab5412ac7f742b8807750acb99b04d1 : SUCCESS in 36m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d35f63e308c451597f519f50779dfa8 : SUCCESS in 32m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/287df240700e481192a98adc4c608f65 : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3932f01d831491282e4c932ac9fcf7b : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8e7bacb3094757aefa2f4520b33128 : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db35ad70284446dd8434dd3398020d65 : SUCCESS in 19m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c6e2eae4fe427c9d2c67456642fab2 : SUCCESS in 48m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/826d668a292c402ebea8dd86ca503e2c : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f28f25c519eb96b38d56f0df823e5aff24bea4b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 00:12:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2ffbdf33a24a6ea695a3a8c480d6f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e590856e03ce481a80b8853eb2aca2f1 : SUCCESS in 1h 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"695eaaecb8cb72206186b0377f2e55ea883ab4c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-06 01:32:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19fb9384cb7111aff753e9cb2d534fa48d5fadcc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-06 15:46:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6bfdd4c0a4d58d6b73c67fc86be4f7f1a61c5cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-06 17:00:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"425f4045979564f04a7c1c2265e36471c46f2ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:20:51.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca445fa326f249beb158cc5ce0c5ed12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966578776d874f83be91e0dee745fac6 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/a5aaca9d75de42f5832cb3afbe588e2c : SUCCESS in 1h 11m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23901ec4e3c74700a8b7aeb8480b30c3 : SUCCESS in 1h 04m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf1604851ee24a3f986c0cc91c6f87ef : SUCCESS in 55m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bd7ae6220ae4baca4b7f3462fda0e1e : SUCCESS in 1h 09m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5ad4dbbf39f4c2ab1c3212299a71069 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15747aa9fa6d4cd4bcca803c8962a270 : SUCCESS in 24m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/362d264b23844241a685c3efb169a116 : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51e3a0dec8874a04923ab669d89c06d5 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f1af832cdbe49e0a67281034c1744f5 : FAILURE in 22m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fd658301010415494ae50be6f088750 : FAILURE in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5f1917575ada428992b5f6dc093ed683 : FAILURE in 26m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/43f4ed0bba9f4ed5a961e346944f5c4e : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/468d5ff6320849afa5a0e434092078e4 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c37705f95ae4ac68868e0b8e6a8801c : SUCCESS in 13m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0379c6ced476491f8a860fa35c239c6e : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c376da60e8fa4b089890b505149ac848 : FAILURE in 18m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ca5abc774934136895724851e27e1b6 : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d49ade413f3f4eea9d829bedecab895b : FAILURE in 58m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/718f3e29f1184c0a9996a62d52e3c114 : SUCCESS in 53m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03e7042528314f31ab641caee893a89b : SUCCESS in 57m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68fe876af17e46998a0c2d3b681f7f88 : FAILURE in 22m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24ddb6fadfd4787a46a52d715ed9461 : FAILURE in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd319d3c2293431691d116a371b63fcf : SUCCESS in 19m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48ad78c56661414cad5ac74827717161 : SUCCESS in 49m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/811f669cc6cb4d9caae3d3626c1dc5d5 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5aeeac29eaea1088a534933dbf7b677c266480bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 19:46:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3a8695ad20419e8664e8e3d15ae741\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbaa3caea15145e5b59ea1612d6b6ead : FAILURE in 1h 33m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d84158ffaf6ab57fa0a67d5ca56117c40d0850b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-06 22:08:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"979ee75239a9ba020f037f984dea233faf8e4605","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-06 22:25:01.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"20d49e03f18cebe18ff07f6c82e8050d9d101e49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 23:17:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b6a840ebb1d4e71b6a9180ff49c5a40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e565942d22994fb38d28673cbf5b6065 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d85fe8d4efc84a98bf52c354bc8674d0 : SUCCESS in 43m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aba119953344440986b86bbf63771043 : SUCCESS in 43m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/131fbfdea90449b79d13ed7597798463 : SUCCESS in 42m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d489c4c94bf047b5847998670d4ec915 : SUCCESS in 47m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad2827b537624c66b5a0b3d925795aba : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b32bf8e46be47f6bbbf89e84e4b88f9 : SUCCESS in 12m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9972fe5725e840f18672ce4fb528877d : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90978c8544ba4aaab2bc6d9db35c534f : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5638b9796b6541358e3ed86071e05bd3 : FAILURE in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c27dea5ab11348f1b35103bef03bfa84 : FAILURE in 19m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2fa13b6984244cdbb4f5d191eec96232 : FAILURE in 18m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/727679bfd4a4447c94f9fd3c5692d5ba : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/331edb2c0a06445fb5c26c92f283c07f : SUCCESS in 9m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa1aa0013fa64c7bbe653b7c06374cdb : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c0e0c7d1b2746f9b83dad95c49f9be3 : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e48c86518fc949d3b0b49f0e5d9f9bea : FAILURE in 19m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/096bbfee981843a9997f64a392143aa1 : SUCCESS in 5m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b02ce587b5d74474933a6bd36be23159 : FAILURE in 1h 00m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e8f5400c00c422ebd767df74600da0e : SUCCESS in 50m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8a19356a6d64aff84d30abe8c850d45 : SUCCESS in 46m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbe58dfd0755443390a9fda8b3401e85 : FAILURE in 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eaf44b4abc640ecaccb89e5edc13761 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9740b268fb904bdab29ea03fcecfde4e : SUCCESS in 25m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3fa2c7abd114ea6b9963c81aa1da4cc : SUCCESS in 49m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48cbac83234d463c95f9baa3a5b63876 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":3},{"id":"47968b5f26af3b555a531d82dadc018c432b3952","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 23:54:37.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5652da208b274438b16084f1f2610049\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b310761665774137919452daa4c6a423 : FAILURE in 1h 43m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c63f3a73b73e7395073d6ed5c6d6b2cccc7a48e4","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-03 20:02:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e577c38f99b482653e1732094300af5f7a92bde1","revisions":{"57782c24266abd2c85c177475000e51c15fe82b0":{"kind":"REWORK","_number":1,"created":"2025-02-05 22:13:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/40/940840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/940840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/940840/1"}}},"commit":{"parents":[{"commit":"d125634b3ee8795b3ad9a2c017c60753d94ec08c","subject":"py3: Fix unicode-header-name handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d125634b3ee8795b3ad9a2c017c60753d94ec08c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-05 22:12:59.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-05 22:12:59.000000000","tz":-360},"subject":"sq: what is a lower normalized meta key anyway?","message":"sq: what is a lower normalized meta key anyway?\n\nChange-Id: If29c420c02f375152e6292c67d0d9b49a4a8e677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57782c24266abd2c85c177475000e51c15fe82b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57782c24266abd2c85c177475000e51c15fe82b0"}]},"branch":"refs/heads/master"},"e821e4682d82b25d53cdf61fc72af7e5a9ec93e5":{"kind":"REWORK","_number":2,"created":"2025-02-06 17:00:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/40/940840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/940840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/940840/2"}}},"commit":{"parents":[{"commit":"847e5dd9965db9bd3a35bd2be5d87e0eaa797eff","subject":"sq: just fix buffered http \u0026 the probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/847e5dd9965db9bd3a35bd2be5d87e0eaa797eff"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-05 22:12:59.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-06 16:57:08.000000000","tz":-360},"subject":"WIP: Normalize HTTP headers consistently","message":"WIP: Normalize HTTP headers consistently\n\nNeeds more tests and a more exahustive audit.\n\nChange-Id: If29c420c02f375152e6292c67d0d9b49a4a8e677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e821e4682d82b25d53cdf61fc72af7e5a9ec93e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e821e4682d82b25d53cdf61fc72af7e5a9ec93e5"}]},"branch":"refs/heads/master"},"e577c38f99b482653e1732094300af5f7a92bde1":{"kind":"REWORK","_number":3,"created":"2025-02-06 22:08:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/40/940840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/940840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/940840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/940840/3"}}},"commit":{"parents":[{"commit":"847e5dd9965db9bd3a35bd2be5d87e0eaa797eff","subject":"sq: just fix buffered http \u0026 the probe test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/847e5dd9965db9bd3a35bd2be5d87e0eaa797eff"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-05 22:12:59.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-06 22:06:11.000000000","tz":-360},"subject":"WIP: Normalize HTTP headers consistently","message":"WIP: Normalize HTTP headers consistently\n\nNeeds more tests and a more exahustive audit.\n\nChange-Id: If29c420c02f375152e6292c67d0d9b49a4a8e677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e577c38f99b482653e1732094300af5f7a92bde1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e577c38f99b482653e1732094300af5f7a92bde1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
