)]}'
{"id":"openstack%2Fswift~941142","triplet_id":"openstack%2Fswift~master~I9a793e4c352cd40a073326315a6ae144a14de1e0","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-11 22:54:32.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-11 20:12:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I9a793e4c352cd40a073326315a6ae144a14de1e0","subject":"Remove some dead code for older pythons","status":"MERGED","created":"2025-02-10 18:13:38.000000000","updated":"2025-02-11 22:55:39.000000000","submitted":"2025-02-11 22:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"941142","meta_rev_id":"2027225cde91f94bec1bdd4cf8ca01c7bd1559ea","_number":941142,"virtual_id_number":941142,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 22:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 20:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 20:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-10 20:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-10 21:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 10:11:12.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"a568d14c819b998f62fd03ad9334eca8340f44c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-10 18:13:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e670e4a9c7726db1583c51fdd895124f8951f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 20:52:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5bcc62dd5e4b3594bcc084e559dfd2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2854dd9efdc44365bfcee69570623a21 : SUCCESS in 2h 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b6829da7c888173d3c3b1b02e92de257de5ed272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:34:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bba35b85c3e495d818535c9809a5940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb452ad8dbc246a3a32108e7e532e6f8 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4e454671ab8d4b1884760998ac67443d : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9939812c42fc49be9770da8af9a605a2 : SUCCESS in 1h 02m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80ee1eeb55904756b2913f2e274c5546 : SUCCESS in 57m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b397d9301e44405aa0acf4d0cea3028c : SUCCESS in 1h 05m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/051c1d30b1624a9ea95cf9805c168e7e : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d7ba32b02d8402da52af5e231cd1cee : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48bf8e345cc640959f96743d2009b249 : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7ff9beb44854448b332a44eed764d2a : FAILURE in 5m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/406dea20d4fc4ce09c34cf54a71dad3c : SUCCESS in 19m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/beabf95971744437a7fb11ef84e18c8a : SUCCESS in 17m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57b6f09ebca44b6594377d7e83b37cd1 : SUCCESS in 35m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0eba72e4ad0d4946b67723c163dc5e0d : SUCCESS in 11m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f21e6139c96a46d0bc24f1d1bd709ae4 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3836bd360f15411aa964cb6ee81b6053 : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63205d227cb649d79fdaec6aa02f214f : SUCCESS in 8m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72c11fc14d6544b296369730b94f473a : FAILURE in 15m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8053501e71a54bc8a138251dacc1dd7e : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cf3864bf6104f42b3e79ef41cb8fbd6 : SUCCESS in 58m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef8c1ce324e64b18a47e522dc6fd63a2 : SUCCESS in 48m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68de58aa7b7f4a7a878b0cb39cfe39ff : SUCCESS in 34m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d786712ce9425e8211ba10963b953f : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d078b6291049b4a66f41c3a39b50a2 : FAILURE in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4dbad8e096b42b6912d4e4d931a31ed : SUCCESS in 19m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cea42d7d4dd04d26bbdc4809bc75905c : SUCCESS in 54m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13a54e577cf84c179a05a1cee5b2ba51 : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":1},{"id":"29eefd06e86083724a8bdbca6626958c44c57f40","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-11 10:11:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41e954059e940ef228dfe94cf96d56a056a0a0c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 17:02:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc2400ffee97a8af740440da724e8d9815846d18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:14:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6362d678255450c86c2e620b5198cbf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cda9e04951b349a19ee67b4e4f8bf333 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":2},{"id":"966110b4ce2cfec80cb3f637a50d2bfc3d348c62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 17:45:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1646a615aa4bbbb264f8447beca5de1e8158ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 18:39:26.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8ef1d561afa41e6a64b3a40bf25f707\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcbf5cc7dc234c3eb2cc05de6076c9fb : FAILURE in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/6e14e76a657d4d59a071797cae44088f : FAILURE in 48m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8ad681654cb4ca6bb667466b52880f0 : FAILURE in 53m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1463e53a8a04592bda50d5b60eebc07 : FAILURE in 19m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f76b6e9b98164c0aba4ece044a036010 : POST_FAILURE in 27m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5a5de0f9533400d8c863100d8c95a55 : FAILURE in 7m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a404912f4ae423f8459b05564d635ad : FAILURE in 6m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46ac51629e3e456183fd55fbe33ec4f3 : FAILURE in 5m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/949396ebe327455796530f92a344b624 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4259ecf7d044dfa81903be909ae47b9 : FAILURE in 4m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/992ce50d0e3a42b9960492b54924e2dd : FAILURE in 4m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/111e297326a649e7add9007b5bf84ea0 : FAILURE in 7m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c7cc272dbe4c41d88d40d255ebfca0ab : FAILURE in 4m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/59d0a2390a2b4d4c9d4df1c092e6b65f : FAILURE in 4m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9f5d2e99b5504f54850f5cca14ce636b : FAILURE in 3m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5b21cdffb5b4b0aaa9e589d35ab6c50 : RETRY_LIMIT in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d88a7dab0ff04a0c85f83655fb9148df : RETRY_LIMIT in 3m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe690309d5604324b8e10de6526341d9 : RETRY_LIMIT in 3m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8ed5d9f0c69414699786a6e88a258e9 : RETRY_LIMIT in 4m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8d3fe33bb8e4e24977e223cc898687a : RETRY_LIMIT in 8m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/100b2007c1f343088f0e729427227198 : RETRY_LIMIT in 11m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95500bdcc045426da2716fd17dc0db3e : FAILURE in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b0d607d49643089e1b0e60741fb33b : FAILURE in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6abe2d3cdb44af29f1f2faaa337f036 : TIMED_OUT in 1h 30m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53af5207ccb94a5f8ebdd75313fb48ff : FAILURE in 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f96c7d924c64466d85b2f9abe7631646 : FAILURE in 18m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7e95f7a6dc677f7d138e56824baaf99d86752bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 20:11:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4ed2d9813a09edd1b8acc248619988787845d79a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 20:12:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b7e6f3f451e43a88ba6c38e7378190c76c424efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 21:29:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d291609db394148aefdc164e17878bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eed3829685a94ab4832cf536da19957f : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":3},{"id":"992048bc32dc8a49309e82a7c80363f59ed83df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 21:31:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b8a4210c3c4da09ca06ab5f86001a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b673e06c4a4a59b64a2f8865efa3b0 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/666af71075f548efae4514ec4c3802f3 : SUCCESS in 53m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/864d2791ea28481581edbf2b4a26f108 : SUCCESS in 47m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89843b9274b34438aba3e29a6117f311 : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/287cf0544cff49e3bee3b02097cf2057 : SUCCESS in 57m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52a85edcca7d49d2950c3b2be56a66e0 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/642bf76039c6482fb552c1b44218c175 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba59fe773b554b29854ace3e4c528e3e : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/888fb84b2cc14f39980980eb6b98f548 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6910e3f688784412acbbd3bb582300af : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e259fec2d9e84ae1856363570dd53755 : SUCCESS in 24m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/09033cc9c0ce42a29f4425e9e17e97f4 : SUCCESS in 28m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c738ad270de444c1aa5a418b5eda4497 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97f2258209f441ef8b68ef74b5278244 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1378801456f24e6480a6da3793c421fc : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ed0f4fdfb3e4425b5413a9d47fe5815 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e06006b02ec4a02b04eb5c2a209bad0 : FAILURE in 18m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6631b19f90654c3c9e8d97c4b84a846d : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2155d5e1020b45dd8e7405f70ffff321 : SUCCESS in 57m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c8c9bbd02c64edb97c888af985255bb : SUCCESS in 1h 10m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2067dcfd480146309617c493d2eac844 : SUCCESS in 50m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ce726abbd264ae6ac5db79f617d3925 : SUCCESS in 24m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c42d827814648ab8cd30de736445318 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b74397f6e67b45149458ab80ad257250 : SUCCESS in 21m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c33d81de5c84100bcae7715e2e904cc : SUCCESS in 1h 10m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aaeb26e71ca431da2839ba1f3bc4cb5 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":3},{"id":"dd7387108e09a4a2f127d0dd1da53f7a2675fc95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 21:32:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f082a95d3a07c1042c0c44faac2ded61e032a763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:54:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1528fa5a565467aac1ea08ffc3b9da3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e37fee9b4204d41b91aff8a1ad08c69 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2fe55b3845824e4eb4fed52354cde452 : SUCCESS in 1h 06m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b10ed820c78943998f2d49556845c5b6 : SUCCESS in 1h 15m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcfc994cada2436dae702b2d94395f43 : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22221e4529214b2a9a910a342a7e69ba : SUCCESS in 1h 08m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6a34e88d7bce486688b83227bf8538ba : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae036da448fc45d08a94d99cdba22992 : SUCCESS in 22m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/78460893eccb4b95a97331963505c6c3 : SUCCESS in 16m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bda73fd6a05f499fb273aeb16a8cbd02 : SUCCESS in 28m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b0005eb225b4f9c9cf8e27cca9cf4ce : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2cca5a41ae9f4938b07c65788653fa55 : SUCCESS in 13m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5438bfa1226c488082a0038d6d3dfbbd : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30f83bcd23514b09bfa2a264ea314aef : SUCCESS in 7m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c86eb824e4ab4758b85220da3f6a02cf : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6503a72591b5435181e15c5bf6a5e134 : SUCCESS in 42m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1751190224a54b1c8858bbb7f3e2159f : SUCCESS in 39m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81b1a95a62c646429bed0c73e99202d4 : SUCCESS in 51m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c98f3014da5a4b08bddcde36a33c1133 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c4a5ea5c4742a4afb0fabdeabc20b6 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b18501557feb4c408fbb07e7952f13dc : SUCCESS in 19m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c585d0ab4c49b480234c3d5beaba18 : SUCCESS in 38m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9216cd2c0b554e018a58721d435f61f2 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":3},{"id":"081e445a5b84676d3ed409058dfdd396619b189c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:54:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2027225cde91f94bec1bdd4cf8ca01c7bd1559ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:55:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6eb1cb9d09408c81ee0e297bd6751f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ad5241f6ec4ba9819fea4bfd635c57 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8892d3b220534219af7cddbf03e578cc : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b94d7960f7ac598760e19179c9ee48be2712ff87","revisions":{"bb21fdfe73dcff8a5f5a01b0c845958e594ec2e5":{"kind":"REWORK","_number":1,"created":"2025-02-10 18:13:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/941142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/941142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/941142/1"}}},"commit":{"parents":[{"commit":"73201cc12dbd0a53bb83b0acf4ed432d5a1b3317","subject":"Merge \"Deprecate expirer options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73201cc12dbd0a53bb83b0acf4ed432d5a1b3317"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-10 18:06:10.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-10 18:06:10.000000000","tz":-480},"subject":"Remove some dead code for older pythons","message":"Remove some dead code for older pythons\n\nChange-Id: I9a793e4c352cd40a073326315a6ae144a14de1e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb21fdfe73dcff8a5f5a01b0c845958e594ec2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb21fdfe73dcff8a5f5a01b0c845958e594ec2e5"}]},"branch":"refs/heads/master"},"5041ddfaa42ce8db33374ff4b2483475871caf85":{"kind":"REWORK","_number":2,"created":"2025-02-11 17:02:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/941142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/941142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/941142/2"}}},"commit":{"parents":[{"commit":"364a47b17678779351ae393a8a473d58698ac53f","subject":"Merge \"tests: Fix flaky reconciler test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/364a47b17678779351ae393a8a473d58698ac53f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-10 18:06:10.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-11 17:02:37.000000000","tz":-480},"subject":"Remove some dead code for older pythons","message":"Remove some dead code for older pythons\n\nChange-Id: I9a793e4c352cd40a073326315a6ae144a14de1e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5041ddfaa42ce8db33374ff4b2483475871caf85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5041ddfaa42ce8db33374ff4b2483475871caf85"}]},"branch":"refs/heads/master"},"b94d7960f7ac598760e19179c9ee48be2712ff87":{"kind":"REWORK","_number":3,"created":"2025-02-11 20:11:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/941142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/941142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/941142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/941142/3"}}},"commit":{"parents":[{"commit":"364a47b17678779351ae393a8a473d58698ac53f","subject":"Merge \"tests: Fix flaky reconciler test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/364a47b17678779351ae393a8a473d58698ac53f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-10 18:06:10.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-11 20:11:08.000000000","tz":-480},"subject":"Remove some dead code for older pythons","message":"Remove some dead code for older pythons\n\nChange-Id: I9a793e4c352cd40a073326315a6ae144a14de1e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b94d7960f7ac598760e19179c9ee48be2712ff87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b94d7960f7ac598760e19179c9ee48be2712ff87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
