)]}'
{"id":"openstack%2Fswift~941228","triplet_id":"openstack%2Fswift~master~I871ddcc4ba559e7e4c0d0e28464780c6cd669797","project":"openstack/swift","branch":"master","topic":"p-footers-callback","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-11 19:21:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I871ddcc4ba559e7e4c0d0e28464780c6cd669797","subject":"Migrate encrypter tests to assert FakeSwiftCall attributes","status":"MERGED","created":"2025-02-11 10:28:20.000000000","updated":"2025-02-11 19:25:35.000000000","submitted":"2025-02-11 19:21:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"941228-p-footers-callback","meta_rev_id":"2912e65efdf245a319c6a9f89637f4f88cf8c27b","_number":941228,"virtual_id_number":941228,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 19:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 17:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 17:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 12:12:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 17:37:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"bb45b33a5558597c89cd5f5dbcc21e9d049f7efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-11 10:28:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71081dabdc54655ee1e44ece2c56517313221bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-11 10:36:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c6e18f348f3ee8a23bc9b62af54fb152b8e6a12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:12:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c62cb9bb1e544a8a99689f93a5e81d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cac189a6274ac59464d71b84b97a95 : SUCCESS in 7m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/753cab16a1c94a82b20408db248bb312 : SUCCESS in 1h 05m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0c9b928026748a19a18058318fc7975 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbf08f1f205e4fedb5b0e170b0c885fd : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c1602fb793f4c0a92d6875b136cfdbc : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7432c340773f4281a6766b5085a6f488 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1eaca879554349069191251e3b93d272 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd2d95ffed4146fd8623b16a0d876c2a : SUCCESS in 22m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a9d8d6ef75f141bb845ef94594197560 : SUCCESS in 27m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7fa2f4dd727a46038b7de2667239570a : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a02f1183ac745fc8b24d8232cdb2011 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/83fc0fa325514ce6877b07d00f103f06 : SUCCESS in 16m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f594c950afd4f9e92bd65c68de7b7e3 : SUCCESS in 40m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b65ac5846c2b47c5b182893449ffb1fa : SUCCESS in 36m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/208925562e34417b90176c59e8bc4ef8 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de1de7cbbb104d40a22cf8fef290c3e9 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c682039293fe4d8e910bfdb94837c2b6 : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ab9f3d6a53e52c681f4fcd3c8515e5285760deb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 14:53:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d7bbaebed484720a3d117b19f4c4c54\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc6ac769b04d44b7a7d57b5c682b1399 : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":2},{"id":"10688fa3f5cdba0c4974c8e31e045dc83e57e168","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 17:37:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bc3d8eafe220db6e8646286cbddadf6ad220f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:37:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec5b639757858e048863c69ec575de737c888c61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:21:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81bd48be0ad046619e296f7648e98a66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57bde4bad5ab4795bb2244f6f9df8a8a : SUCCESS in 8m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec3c00c65646420d90a6afe9dff2b820 : SUCCESS in 49m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2ee1ef8452dd48149971119ebdae5aea : POST_FAILURE in 6m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f8619ab9d4664f8cb052917b1e683c8b : SUCCESS in 10m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce97bf5e6e9642f08ce7decda46af336 : SUCCESS in 20m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/27bce778a74c426b9287f17b75bb873d : SUCCESS in 15m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6aa572394c64d1e95ca6e7b10e77bfa : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/817c3d74998e49e082efeef1e2c3099d : SUCCESS in 13m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dbdd4840185e40dbbda0ea942c2d2fc4 : SUCCESS in 10m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d109ec551d74d23be89893d2f47dfed : SUCCESS in 5m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b5ffe4126d624a4eb28fe86c55dde493 : SUCCESS in 7m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d5c3cae62ca42f58c8ab0f26647720a : SUCCESS in 48m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4ba5b99ef2a4edab0b3ab797b02c9f4 : SUCCESS in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bceed2c9ba8040a9a79b84098bb4c61a : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6125d58a71474613b84559528fd8e85b : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cffca70610e4514a6ee250fe3f47036 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c467e7c9a4685bab29130cdfc88ad5439454bd08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:21:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2912e65efdf245a319c6a9f89637f4f88cf8c27b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:25:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df317dc0b71642cbb4de431cfd4939e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df2f526ecc54979b4d4f6bbbed3ffb4 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/13fddf19cfbf4057a563c929c524c892 : FAILURE in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2560eb7cb559e511b0558dd1cbd7b89ec60dda0b","revisions":{"8e9c9d45941b0d4ceeb3ec3e3b15d04a5851253e":{"kind":"REWORK","_number":1,"created":"2025-02-11 10:28:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/941228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/941228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/941228/1"}}},"commit":{"parents":[{"commit":"f6fa84743fb4f15d2df946642c5804cbe630f627","subject":"Tighten up encrypter tests re. request footers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6fa84743fb4f15d2df946642c5804cbe630f627"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-11 09:52:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-11 09:55:59.000000000","tz":0},"subject":"Migrate encrypter tests to assert FakeSwiftCall attributes","message":"Migrate encrypter tests to assert FakeSwiftCall attributes\n\nMake encrypter unit test assertions more explicit by using assertions\non the named attributes of FakeSwiftCall rather than assertions on\nposition-indexed call tuples.\n\nChange-Id: I871ddcc4ba559e7e4c0d0e28464780c6cd669797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e9c9d45941b0d4ceeb3ec3e3b15d04a5851253e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e9c9d45941b0d4ceeb3ec3e3b15d04a5851253e"}]},"branch":"refs/heads/master"},"2560eb7cb559e511b0558dd1cbd7b89ec60dda0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-11 10:36:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/941228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/941228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/941228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/941228/2"}}},"commit":{"parents":[{"commit":"a740591883076c791944e40eed109a521ac38604","subject":"Tighten up encrypter tests re. request footers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a740591883076c791944e40eed109a521ac38604"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-11 09:52:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-11 10:30:57.000000000","tz":0},"subject":"Migrate encrypter tests to assert FakeSwiftCall attributes","message":"Migrate encrypter tests to assert FakeSwiftCall attributes\n\nMake encrypter unit test assertions more explicit by using assertions\non the named attributes of FakeSwiftCall rather than assertions on\nposition-indexed call tuples.\n\nChange-Id: I871ddcc4ba559e7e4c0d0e28464780c6cd669797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2560eb7cb559e511b0558dd1cbd7b89ec60dda0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2560eb7cb559e511b0558dd1cbd7b89ec60dda0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
