)]}'
{"id":"openstack%2Fswift~941365","triplet_id":"openstack%2Fswift~master~I332ce724aa10287800cbec8ca21aacc3bbd3c22a","project":"openstack/swift","branch":"master","topic":"p-fake-swift-encapsulation","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-06 22:22:27.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-06 21:11:58.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I332ce724aa10287800cbec8ca21aacc3bbd3c22a","subject":"FakeSwift: more encapsulation of request properties","status":"MERGED","created":"2025-02-12 11:54:37.000000000","updated":"2025-03-06 22:23:49.000000000","submitted":"2025-03-06 22:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"941365-p-fake-swift-encapsulation","meta_rev_id":"fba066fc47489a658a50f0e39a8108b42e68aca9","_number":941365,"virtual_id_number":941365,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 22:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-03-03 06:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-03-06 21:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-03-06 21:11:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 13:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 14:58:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-03-03 06:12:29.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"c7eacd13447ba5096c84f182f3fd5fad85585ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-12 11:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4eed019b466f9d28f653ad79ef7bda13686e9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 13:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d41f8ef768647ff9a49b8643b103df8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1bebff68154f40b8cf73fa4f493b8d : SUCCESS in 9m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee17e8afde544912b31d440c5f2357c8 : SUCCESS in 49m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d515cdccced8460ab9233fe1cd51fc66 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77a55c0aa06e44609c8f56d6f414ae55 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11cc09879cf24d1fafcea3f8abb9c463 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d13d712c6c114de4ba35bfc4cb493433 : FAILURE in 3m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e9ad5b38abcc43feadff00c60a194943 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a768d44b5bb45f19de2d2ac96416c74 : SUCCESS in 20m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7684055b541f4cac9b8087c7e79ddbc3 : SUCCESS in 28m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5355adb7c1f44fa5948b5c86214cbc70 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ea10a40e37ad4ea394d22a97bec4cbf9 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d62e1471a87140d0a0ee3e37d7300025 : SUCCESS in 16m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c60fc7da53e841eab51af71c9a16ecf9 : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/740af4317476411ab654f9a08158c228 : SUCCESS in 1h 01m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dd3fc1bf9f84eab9becfb3f34be89ef : SUCCESS in 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4595aa965e7841669d244a5a00ea6c4f : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ba0c1ba7613442c980d45d708efe86c : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a0de7c015d0a87e52f161a25eb1742e8bcb25dc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 13:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b431e4b38a2485c9e31b7059f51a1c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/794ac5fd96304210afcc47d1b19c13b9 : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d1e1a18d3037aad82804f7e89342650a993c495a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-12 14:58:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef71fa6a2d2d37d4d56e38e01173d95476e94bd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 11:38:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b7291d0032708dfe3cf2e1645af8b86745873529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 12:51:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183dcdd2a60c479e818a8e2c94ddaa9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013282fc4ef442138d99cac45571de45 : SUCCESS in 9m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77704616a1f94813b4cb100c217cfc50 : SUCCESS in 1h 06m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75b3e9737a944709bce3b46db38c8511 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a3baa8c654c4a84b9142e89b7ca6d34 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e74bb10153f842a99c25fac9901304a4 : SUCCESS in 13m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d5067aaa93d4f29bd9f49dcf36e5187 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0eb4241bcc41bf9750e058eb8ad361 : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62359eb81bbe4760b59f23e0c9280563 : SUCCESS in 19m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/42de9b8c546f4b3baf246ef5453b4e35 : SUCCESS in 27m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dfa2c874e4134e039cf90770ddaa9f3e : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9424f7829a14ea08bcde0ca622c079a : SUCCESS in 15m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5d3896afecc45a0be5ac1d60e046492 : SUCCESS in 16m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c19cf62a24c0493da43be2714737a3c5 : SUCCESS in 42m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7efa665e1f6a4c35a9b7ab86f5593ee6 : SUCCESS in 36m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ad2948f8dd74836af1e6b8b20848371 : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d104b311524d069f06c034139cc531 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50c996227fae4dfd845695e47fc2e8f5 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"950433af355b18696fb19220896a1c9f10454406","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 12:58:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1dbbf074df490fb143f7dcdbbc25d0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/809090c3d1504ff686bdd5407cf32039 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d98ed6c339f821a15395417d18a0b1c5a453ab6f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 14:59:45.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46e0777965eb4abec3dbc8bca86971adcc5a180f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 15:39:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab228da7b9081d3a048016c622330a2999b4a32b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 15:45:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6f271847274811c14f3e874f19fd583adb41dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:57:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3653b0faa68a40628fdbba94bc52df5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7fadbff4c446eb90afbc8daa8a202f : SUCCESS in 10m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16e3877f8fd8490088f14da0c9383873 : SUCCESS in 1h 04m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bde0f7c4e0af456b95f46510c6c66c04 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b731e74b25044b02a67747e2aeb0591d : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b085d45819f64218a13a0a62543b0e56 : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78aa0059057c4006a925986c6f054f70 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16b6008270e440828656d9375e91083d : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f064c78db7c46839219c593b8990a14 : SUCCESS in 20m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2bb86d12424d4688aec760e0c7d38031 : SUCCESS in 35m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/563cbb9b630f47c695c4f4409e951e5c : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/284101a57a5b401183ab7cf388eb1bfc : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c3b925a27a024296a475576acad9f0c9 : SUCCESS in 17m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc179d50fb254ac4b491e61071dc6543 : SUCCESS in 49m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98b8708dc7404342aed97c42403c5824 : SUCCESS in 34m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75fe48d1b7004fa08098ed8d89f60a97 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a33b913fe89482aa8926df88640356e : SUCCESS in 8m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8aac9cda7dd4be69b44ea57151ec756 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c17e65715e0f4ef12ae2db287382ab163d19c93b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 17:06:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe5b20a8bf5841979085fc4e5f281b49\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aaf95a4397ac4eefb18334cf2ac057da : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9412e0368912bf7da5e68059b1707c2129e516cc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-03 06:12:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1473b14bcf5a019c7134ad213a1677268b838562","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-06 21:11:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ddf04efd00f5b23937b4510408b4fd90c61fec5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 21:12:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9713dffe0a7cf459f006d701ab5eee887f90cac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:22:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca383ae7b754894a6ba035adaa6b3b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beffb29dc9be431196355567b6fa8eca : SUCCESS in 8m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dcb29d7c5cb24115b5d06fd656a99847 : SUCCESS in 49m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f0947356749c405592ba11a2adb0447e : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/894c072916fd4630bae4ff83c442fd36 : SUCCESS in 24m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a806831945114279819663c234fc9fc7 : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c25c1f078fa940b398a549b5cad33e5a : SUCCESS in 27m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc9b0c78c15343feb46a7f5087063035 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7df777719e54449995508d4875a50b64 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2e7e62b3fe8c4a8c9e93d2c13c4b8ab9 : SUCCESS in 19m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d2bc4eda8b145c29201fae96e2099f2 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecb365e88a7b49319c65b4d785b3b126 : SUCCESS in 6m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1df58d47019c461996d08a1aa4e128af : SUCCESS in 52m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07cea4cf61a84e08a4bf2eb6b2dba32f : SUCCESS in 35m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cfffa5c81ca435f99a432e97507dffa : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e526fa4cc4dc4d54a3445890a7b36322 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/845c31d52fca4391887329e8d270be00 : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":3},{"id":"05b01a7df0c2a30d30035abe1287adbe070e7260","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:22:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fba066fc47489a658a50f0e39a8108b42e68aca9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:23:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18978ffee0c642e4af835121b48bdc23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2edf3d0b44994a8fadd536e8acfee4b8 : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/008ab65d5bc34d7b809d9e163f9f3218 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ceff47d7af9079c5c815d83663b3ab12eba23e60","revisions":{"4a9a8b6d28a677380e93a3d06e405b7d1569fdcd":{"kind":"REWORK","_number":1,"created":"2025-02-12 11:54:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/941365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/941365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/941365/1"}}},"commit":{"parents":[{"commit":"747b2a4a328b62585444de37f96921b1401bf565","subject":"Merge \"Add per-container storage policy to account listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747b2a4a328b62585444de37f96921b1401bf565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-12 11:23:13.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-12 11:52:56.000000000","tz":0},"subject":"FakeSwift: more encapsulation of request properties","message":"FakeSwift: more encapsulation of request properties\n\nCapture relevant items from the request environ as part of\nFakeSwiftCall and use that to query captured txn_ids, sources rather\nthan maintaining separate lists for each request property.\n\nStore captured request body as an attribute of FakeSwiftCall to\nsimilarly avoid maintaining another list. The request body is read\nafter the FakeSwiftCall has been captured in case there is an error\nwhile reading.\n\nDrive-by: drop footers arg from FakeSwiftCall constructor since it is\nnever used, nor should it be used.\n\nRelated-Change: If24b6fa50f1d67a7bbbf9a1794c70d37c41971f7\nChange-Id: I332ce724aa10287800cbec8ca21aacc3bbd3c22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a9a8b6d28a677380e93a3d06e405b7d1569fdcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a9a8b6d28a677380e93a3d06e405b7d1569fdcd"}]},"branch":"refs/heads/master"},"11eee8741ca16cd25fcc2feb704d7c2340750377":{"kind":"REWORK","_number":2,"created":"2025-02-13 11:38:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/941365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/941365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/941365/2"}}},"commit":{"parents":[{"commit":"747b2a4a328b62585444de37f96921b1401bf565","subject":"Merge \"Add per-container storage policy to account listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747b2a4a328b62585444de37f96921b1401bf565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-12 11:23:13.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 11:21:00.000000000","tz":0},"subject":"FakeSwift: more encapsulation of request properties","message":"FakeSwift: more encapsulation of request properties\n\nCapture a copy of the request environ as part of FakeSwiftCall and use\nthat to query captured txn_ids, sources rather than maintaining\nseparate lists for each request property.\n\nStore captured request body as an attribute of FakeSwiftCall to\nsimilarly avoid maintaining another list. The request body is read\nafter the FakeSwiftCall has been captured in case there is an error\nwhile reading.\n\nDrive-by: drop footers arg from FakeSwiftCall constructor since it is\nnever used, nor should it be used.\n\nRelated-Change: If24b6fa50f1d67a7bbbf9a1794c70d37c41971f7\nChange-Id: I332ce724aa10287800cbec8ca21aacc3bbd3c22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11eee8741ca16cd25fcc2feb704d7c2340750377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11eee8741ca16cd25fcc2feb704d7c2340750377"}]},"branch":"refs/heads/master"},"ceff47d7af9079c5c815d83663b3ab12eba23e60":{"kind":"REWORK","_number":3,"created":"2025-02-13 15:45:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/941365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/941365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/941365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/941365/3"}}},"commit":{"parents":[{"commit":"747b2a4a328b62585444de37f96921b1401bf565","subject":"Merge \"Add per-container storage policy to account listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747b2a4a328b62585444de37f96921b1401bf565"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-12 11:23:13.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 15:45:47.000000000","tz":0},"subject":"FakeSwift: more encapsulation of request properties","message":"FakeSwift: more encapsulation of request properties\n\nCapture a copy of the request environ as part of FakeSwiftCall and use\nthat to query captured txn_ids, sources rather than maintaining\nseparate lists for each request property.\n\nStore captured request body as an attribute of FakeSwiftCall to\nsimilarly avoid maintaining another list. The request body is read\nafter the FakeSwiftCall has been captured in case there is an error\nwhile reading.\n\nDrive-by: drop footers arg from FakeSwiftCall constructor since it is\nnever used, nor should it be used.\n\nRelated-Change: If24b6fa50f1d67a7bbbf9a1794c70d37c41971f7\nChange-Id: I332ce724aa10287800cbec8ca21aacc3bbd3c22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceff47d7af9079c5c815d83663b3ab12eba23e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceff47d7af9079c5c815d83663b3ab12eba23e60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
