)]}'
{"id":"openstack%2Fswift~941475","triplet_id":"openstack%2Fswift~master~I7f0a89916a45e7750c75022c6e7b20f3a3ac610a","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-13 01:04:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7f0a89916a45e7750c75022c6e7b20f3a3ac610a","subject":"sq? share validate_sha256 and other fixups","status":"ABANDONED","created":"2025-02-12 18:56:54.000000000","updated":"2025-02-13 01:04:30.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"f9d070f2c3c7bd744df5b6e1bc326226555342d2","_number":941475,"virtual_id_number":941475,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-12 20:22:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-12 22:53:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 20:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 20:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 22:53:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4a4c21d30fe29ba2e9260707f3f2e7472c3bd0dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-12 18:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b0a054531babb7184b5be0e28a9b0b30df449ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 20:16:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ccff155f4f0427080363c632ee42be6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca33ad4fc95f4e7c924b8bdce5c2aec4 : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4ebffeb69da965b26184e58b22b2bc501334e144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 20:22:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985e17a662664563ad5dfcbff4c6bd20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c772a588094281a716005d922f92d4 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/1a6743e84c6b4da198b5531a815dedee : SUCCESS in 1h 00m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2447fbbb01e496c83469ef6459e599b : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfec9317489946a3bb7d5bbd327ac86a : SUCCESS in 43m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0823d8fc4bed4688918835f6349579da : SUCCESS in 1h 08m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e431df8a9e8b41bcadd701ce295fa86a : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a412d01b9e74b828181eff1a69c4075 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3097acb33dfe425387d73eaf7c553326 : SUCCESS in 18m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b393230096547938693a2881bbb2102 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e91c84886c64cbb9e73ab340780ed86 : SUCCESS in 14m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bbbb40205a694c938b7b2ba6d42f1544 : SUCCESS in 20m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9ca60eca98c443ca0b3e5277a191b30 : SUCCESS in 22m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a086cd4b462467886d1927e10ae725b : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fb97a30fe77d4c0fa34d099051240385 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ddb035949d140ceac969b0444627dda : SUCCESS in 13m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4501d767d25f43239c2e9f186d4101de : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2da37364338f4637b20f22e491e0d1cb : FAILURE in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecaa7b6ba66040269e421529bb066de0 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4baf82e8a51f4fa5a346a94d1938b2b3 : SUCCESS in 51m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1657b0d52754d11b830c94f8232033f : SUCCESS in 45m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8eaaa955c56d4031b30a6191d772ca0f : SUCCESS in 51m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02ae9da79f11465fadfb66569a3e32b7 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826dad465569490199eeab686dd66009 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e6028faa0054a7e8bfff7374b4e1561 : SUCCESS in 12m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6ee7c66de4d45419ce580e372c368bf : SUCCESS in 38m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dc39249c0304d6681568d5caf4e6398 : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":1},{"id":"247070ccc69207b88e6fa7d459fe3b6ee9188ec7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-12 22:53:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9d070f2c3c7bd744df5b6e1bc326226555342d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 01:04:30.000000000","message":"Abandoned\n\nSquashed into parent (mostly)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad9e2cffea8836db92e9fc1ee040eefd4b95dc64","revisions":{"ad9e2cffea8836db92e9fc1ee040eefd4b95dc64":{"kind":"REWORK","_number":1,"created":"2025-02-12 18:56:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/941475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/941475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/941475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/941475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/941475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/941475/1"}}},"commit":{"parents":[{"commit":"7539776b363b86f33e902c6da61844a9ac72291b","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7539776b363b86f33e902c6da61844a9ac72291b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-11 12:53:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-12 18:56:15.000000000","tz":0},"subject":"sq? share validate_sha256 and other fixups","message":"sq? share validate_sha256 and other fixups\n\nnot perfect but maybe a little better?\n\nChange-Id: I7f0a89916a45e7750c75022c6e7b20f3a3ac610a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9e2cffea8836db92e9fc1ee040eefd4b95dc64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9e2cffea8836db92e9fc1ee040eefd4b95dc64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
