)]}'
{"id":"openstack%2Fswift~941557","triplet_id":"openstack%2Fswift~master~I8732c510c3e11adbfdc71eea16e951af51fc3d8c","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-13 18:27:58.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-13 18:27:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I8732c510c3e11adbfdc71eea16e951af51fc3d8c","subject":"sq? refactor signature checking to classes","status":"ABANDONED","created":"2025-02-13 15:00:54.000000000","updated":"2025-02-13 18:27:58.000000000","total_comment_count":17,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"530afa267c91519408242a97de7d79901bd00bfb","_number":941557,"virtual_id_number":941557,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-13 16:25:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-13 16:44:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 16:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-13 16:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-13 16:44:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b15c1cf6729ee67bbf5f291bb861b0fc8770b2df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 15:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef9bbadcb058dcfafc091b52273cbfcbfd16056c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 15:11:03.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e96aff5e84cd8b5962b645fb763954d3a7b335eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0331471638be4a859e46d67bd729aa6c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/418dafe9807c41dc8be9a959b859a35d : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6f3451de0adc640d1d4cbda9875c6e2abc307cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab74a894282d42b0b6773a13c87159de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa0eb3c7fa34c3a8a27640c12d42904 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ba2501f606ed42adb0dab8f78911feee : SUCCESS in 1h 15m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ffa8d4f8501433fbc93f17743fdb228 : SUCCESS in 56m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/563ec8e694fb478487797984873c4356 : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6622f14dab647d9a8f44f373b26036f : SUCCESS in 1h 06m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adb3f153b74045b1819223579cf68b4d : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b19fac8f8044a0ca37c6f6ced3b19ec : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75e0e031cda348c5ab29ae53e3f73e8a : SUCCESS in 24m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8115b1b646a24938a8ee7c5d0c3b675e : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a82f971a933e4a5c9f28a3fee2ba2b3f : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/86bfb0a0bf584767b23b8ed26cdbd0f0 : SUCCESS in 9m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa209e6137042f5a3a6aa77bf852318 : SUCCESS in 22m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/955f051ee4374d969549af75d607ffc6 : SUCCESS in 12m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1eab9bb181dd453dbdb5a0ec9dae064b : SUCCESS in 16m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3efaff1ea7dd4e4aac99bfed77fc1eb6 : SUCCESS in 9m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c6b726019b44f4e8c9bfa2610ddec2e : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b86fbdebace4f01a7b7f5348954d4bd : FAILURE in 17m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea4899de74384a2684f2ae9a32db30ab : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fda60762af5f4f90a235525fbd9c1397 : SUCCESS in 48m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52a2ac1400314bf3a6a36eb5c41e6d54 : SUCCESS in 50m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f25a601cec644cf6a51adb55de062b47 : SUCCESS in 39m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15b108bfc91e4eddada86605451d6831 : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1aad93c74546aa8c99b71f3cdb5cc6 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b3b18bd930e413c805766174bd38534 : SUCCESS in 16m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c94f301fa7f4443badda19a583daa6d : SUCCESS in 1h 00m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c7d0b8303748e7ad48ca5f7091cbe0 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a63e559b08539582f079afdf2a60f987a94bfa87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 16:44:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5989304d190b2bc4b96509a878f570ef438ae2ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 16:57:54.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3319c6b5c2ee9f02f543d7cc2fdc766c516fe7c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 18:27:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"530afa267c91519408242a97de7d79901bd00bfb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 18:27:58.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9e5b27c6361dba72a98c746aed1d170c3fb2826","revisions":{"c9e5b27c6361dba72a98c746aed1d170c3fb2826":{"kind":"REWORK","_number":1,"created":"2025-02-13 15:00:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/941557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/941557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/941557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/941557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/941557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/941557/1"}}},"commit":{"parents":[{"commit":"cbfce216643ee593efc0f1e72c1087b48bc1a4cf","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfce216643ee593efc0f1e72c1087b48bc1a4cf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 14:52:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 15:00:44.000000000","tz":0},"subject":"sq? refactor signature checking to classes","message":"sq? refactor signature checking to classes\n\nChange-Id: I8732c510c3e11adbfdc71eea16e951af51fc3d8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e5b27c6361dba72a98c746aed1d170c3fb2826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e5b27c6361dba72a98c746aed1d170c3fb2826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
