)]}'
{"id":"openstack%2Fswift~941588","triplet_id":"openstack%2Fswift~master~I5ccf22d770ab9803d7f2dc30d2b09863c5a8ff70","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-14 09:55:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ccf22d770ab9803d7f2dc30d2b09863c5a8ff70","subject":"sq? unit test for unsigned_payload_trailer","status":"ABANDONED","created":"2025-02-13 18:49:55.000000000","updated":"2025-02-14 09:55:41.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"eb90e13fa08b26e54a102cd552f48a13500e514d","_number":941588,"virtual_id_number":941588,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-13 20:11:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 20:11:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 01:06:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"2821be0540878e0b3ac48038e75fea7de808935f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 18:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcc36d5a31d2e41f4e4eebf12285168b09ca88cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 19:18:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"127f4613754e9d3cfba8941e1b43780fb77b2d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:11:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1b6a7843d0491a8aeeac3b404ddc5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76cbc5e2e78d42aebc1f36b905352d9d : SUCCESS in 9m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf9c4dd2a206403d844f6ae294eb9ffa : SUCCESS in 50m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ffd8c8355484c539839549f295ba4d8 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/663e3136ad8f4f448976300486350094 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05fd6df7266846e8912e07a24d468d6a : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c9f4d18321b4ae7aaa22057bb9255a6 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fdb750243c9945e6833715e59c427e6c : SUCCESS in 20m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f145b855636641929b61b4034f80a01d : SUCCESS in 20m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/070edd60a351443a8e5103c096ce5a94 : SUCCESS in 24m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3998b060af3e4026b35ed10de944a6da : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d69023222414db3bbbf14e04b962dfe : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c44c9f1803f249a9a27251f4ef25fb80 : SUCCESS in 19m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f04823d5cd7d4b12bb083b9c18cde6f0 : SUCCESS in 50m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c915261ed2be4b66897a5019883940b8 : SUCCESS in 26m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5096e070ffb54cc8b14bfe8e485837d7 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff896861e0de4fd59e37bdf9b5e66683 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbb3f733b3074060a519ef101404b68f : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"15d212759299a0b0a0428817966e7b1bc1a13d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:43:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d95391e5c4284d0487d3228e15e87c3e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2002973dfab6449782bc775befb942b4 : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"deb024a63047ac2115f45c38fd2b1b3bc6e98ce4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 01:06:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb90e13fa08b26e54a102cd552f48a13500e514d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-14 09:55:41.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8235b9509a8677841f0a866deb392ccf3dee9c1d","revisions":{"3f9b4ba987588d729b03fdf0082b3cfe9bc8a000":{"kind":"REWORK","_number":1,"created":"2025-02-13 18:49:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/941588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/941588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/941588/1"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 18:49:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 18:49:46.000000000","tz":0},"subject":"sq? unit test for unsigned_payload_trailer","message":"sq? unit test for unsigned_payload_trailer\n\nChange-Id: I5ccf22d770ab9803d7f2dc30d2b09863c5a8ff70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f9b4ba987588d729b03fdf0082b3cfe9bc8a000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f9b4ba987588d729b03fdf0082b3cfe9bc8a000"}]},"branch":"refs/heads/master"},"8235b9509a8677841f0a866deb392ccf3dee9c1d":{"kind":"REWORK","_number":2,"created":"2025-02-13 19:18:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/88/941588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/941588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/941588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/941588/2"}}},"commit":{"parents":[{"commit":"1b38348be4bb37c9e335d794b7e52eaff1e9edfb","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 18:49:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 19:18:22.000000000","tz":0},"subject":"sq? unit test for unsigned_payload_trailer","message":"sq? unit test for unsigned_payload_trailer\n\nChange-Id: I5ccf22d770ab9803d7f2dc30d2b09863c5a8ff70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8235b9509a8677841f0a866deb392ccf3dee9c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8235b9509a8677841f0a866deb392ccf3dee9c1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
