)]}'
{"id":"openstack%2Fswift~941698","triplet_id":"openstack%2Fswift~master~Ife2167296648afd663af14336a9df623423e793c","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-17 11:04:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ife2167296648afd663af14336a9df623423e793c","subject":"sq? return NotImplemented if auth doesn\u0027t set secret","status":"ABANDONED","created":"2025-02-14 17:53:24.000000000","updated":"2025-02-17 18:06:36.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8a7f667e6c6de9d12a7e7813b7d000bc7c19e836","_number":941698,"virtual_id_number":941698,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-14 19:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 17:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-14 19:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 19:24:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c13e8376b0ede3b8a9374d10dba9ad7f17e9ee86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-14 17:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d64c7b8528b2209653e1f6f42fb3e9832830210f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:53:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4a0bdba9504e1182ca25a16d609625\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e870b3d938e843f091200d2978bf3711 : NODE_FAILURE Node(set) request 300-0026341371 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6bc4085e7ae96989d6ad72a690703a76526fe2b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 19:03:21.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69b00f9af2584e9e9f59d1fb86d66ac2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ae04d7c4664f7a906d1ad43e8cc9de : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/0799d050fd3a409fb551d1329ced7938 : SUCCESS in 56m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd0341701ebc47ea8a293c1d38c52f68 : FAILURE in 50m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e03c37ec8ea34d12a59e1edcf74c0f95 : SUCCESS in 55m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f12e284da2a64e1f95151562af46eb1e : SUCCESS in 1h 04m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afea472e3a0d45cfb3fef50f774294a3 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3451e1627d64b59bff99be67ec95f4f : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a851f20008e242aba3504e8c53f2f783 : SUCCESS in 24m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4b3d29e805a4e0ca9430a3e2992a957 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee813f871ece427f94f3a635709424cf : SUCCESS in 23m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48179d1e884b4b96818889f5ef316645 : SUCCESS in 20m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40affbacfa7d4ae592c8fa74a3fc1fa7 : SUCCESS in 26m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83edb864ec3c497eb99ea4a6ba70eb42 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29a87abc6f18491caba0d88700167776 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/060ee41463e44a7aaec3eb57b32d004b : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/537b130e48d64242aaca644c1ce29ffb : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c26c37043a34b2888afb0f8bf4e116f : FAILURE in 14m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d350a0f182c4dbd830ff968ff88a8a8 : SUCCESS in 10m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e70f9e4f8e0041f68b4f434805954620 : SUCCESS in 50m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f5e90512cda4cc48ebc5d9d6d8e1d52 : SUCCESS in 45m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a39696240324939abf7b19be2a4025c : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78eb1aeed3a44895b7584f170bb55d37 : SUCCESS in 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6634c824204a0e8501fa8fbf863225 : FAILURE in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60cd0459ff7b47d0b0e5c3e6519c573a : SUCCESS in 17m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/143ab8c0fb4d406ba3702dd133c04255 : SUCCESS in 54m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e50ac9981f143d2aa81e2e7ff09d7db : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0038348191819937dc35d15ff2047935a7e3e846","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 19:24:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b9e48d2db1a8f934cd52515e581bfe3ca97be7a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-17 11:04:23.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1},{"id":"8a7f667e6c6de9d12a7e7813b7d000bc7c19e836","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-17 18:06:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0047cb40a3e3f0286c0590315cd8c2143bea0eae","revisions":{"0047cb40a3e3f0286c0590315cd8c2143bea0eae":{"kind":"REWORK","_number":1,"created":"2025-02-14 17:53:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/98/941698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/941698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/941698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/941698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/941698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/941698/1"}}},"commit":{"parents":[{"commit":"2810ceeb4ed0ac60972d1e5141276ae0236de33e","subject":"sq? unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2810ceeb4ed0ac60972d1e5141276ae0236de33e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-14 11:51:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-14 11:55:51.000000000","tz":0},"subject":"sq? return NotImplemented if auth doesn\u0027t set secret","message":"sq? return NotImplemented if auth doesn\u0027t set secret\n\nChange-Id: Ife2167296648afd663af14336a9df623423e793c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0047cb40a3e3f0286c0590315cd8c2143bea0eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0047cb40a3e3f0286c0590315cd8c2143bea0eae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
