)]}'
{"id":"openstack%2Fswift~941700","triplet_id":"openstack%2Fswift~master~I26121a84aba649d70c77278b45eb0905d9e4f4c6","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-17 10:58:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I26121a84aba649d70c77278b45eb0905d9e4f4c6","subject":"sq? docstring fixes for s3request.py","status":"ABANDONED","created":"2025-02-14 17:53:24.000000000","updated":"2025-02-17 10:58:52.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"1b6a00cc22a20f0f8173cd9c3aeaf387f483240d","_number":941700,"virtual_id_number":941700,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-14 19:11:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 17:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-14 19:01:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-02-14 19:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3ef03aa1c87913217b5e50bd9a602c750374e6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-14 17:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca4417a279b7656cb34b23c4d2bc0f7efc82d6fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:53:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc8f55fd02243d2b1a4c082a86e9e2a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fba2c3fd5204150ace708caaef6fa2b : NODE_FAILURE Node(set) request 300-0026341373 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e465f3ad1fa479325a5e150d42be0fac2906df7f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 19:01:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7755c1a5b12f7564f9725db1d1146adae21b2c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 19:11:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59f7ee46d2e346eeb925c1ed4c9393c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8e2db23a3b4b1897bed161dff21e7a : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/2452e8bb132548b19c02ef3848fb74bf : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ee98f20cb4541ba8e268e4410489a1e : SUCCESS in 1h 08m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af72325a0a374e1e9ff960a303435342 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba479803b21f4f8e8deb53b1bc88f5e6 : SUCCESS in 53m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28fd93ff31e04e25ab1fe4d1c8e7059e : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/363918dbe6d1491ca0f623d4fed5c4ae : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/343d1fb2e2d54b27b902d94f0d1cac16 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/807898689e1149b7af4389031a9e58a2 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c00740e091c04cc89bfb7c913fb8cb47 : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5684aaf68e6e49d79a4ad7c4118b407d : SUCCESS in 16m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c1be6582a7a472abe5d5a8c5a8444d4 : SUCCESS in 30m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f2fe0df438e4b6883d1c8f786e58431 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fe20b35f28aa449d9221e894f5a92029 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30b3f57cc5dd49a4a3f04a79163703c4 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b313b42ac70542c38a7b7804ca3e2f77 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7479a52138f8406d81b5c4f323b9c216 : FAILURE in 13m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1f2da0b32b548e288a773b022d5a5be : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ae4ab547c204572b7033bc2ed3e84a9 : SUCCESS in 53m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/229bccfb75a54781bd694aef1262fc9e : SUCCESS in 50m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96d6f431345743af962893b8dcb2dd94 : SUCCESS in 54m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae886cfe35e7483596faa9772ad3a2c7 : SUCCESS in 22m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fefc515d309149cd88d09425b7c162d5 : FAILURE in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97b03dae0f3c45e3b4434bc105e42ff6 : SUCCESS in 20m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75ddfe9ca2424908ab55cdb346512c1f : SUCCESS in 51m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b680e7c901be4c2893366aad859a1428 : SUCCESS in 1h 04m 11s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/s3api/test_s3request.py","accounts_in_message":[],"_revision_number":1},{"id":"1b6a00cc22a20f0f8173cd9c3aeaf387f483240d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-17 10:58:52.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27c6e73046f7df9c95b8a03ab7aef6abe9993ab0","revisions":{"27c6e73046f7df9c95b8a03ab7aef6abe9993ab0":{"kind":"REWORK","_number":1,"created":"2025-02-14 17:53:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/00/941700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/941700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/941700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/941700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/941700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/941700/1"}}},"commit":{"parents":[{"commit":"602e55197f0352281eaf560208e0087bbfaf97b2","subject":"sq? compat test fixups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/602e55197f0352281eaf560208e0087bbfaf97b2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-14 17:50:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-14 17:50:30.000000000","tz":0},"subject":"sq? docstring fixes for s3request.py","message":"sq? docstring fixes for s3request.py\n\nChange-Id: I26121a84aba649d70c77278b45eb0905d9e4f4c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c6e73046f7df9c95b8a03ab7aef6abe9993ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c6e73046f7df9c95b8a03ab7aef6abe9993ab0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
