)]}'
{"id":"openstack%2Fswift~941744","triplet_id":"openstack%2Fswift~master~Ibe62d567caa4f7c495a222e539090db9aff0cbeb","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-18 18:38:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibe62d567caa4f7c495a222e539090db9aff0cbeb","subject":"fixup! fledgling boto3 tests in functional/s3api/test_obj.py","status":"ABANDONED","created":"2025-02-14 22:53:16.000000000","updated":"2025-02-18 18:38:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9514e32f29eb3c8dd27f5a4a1a3492fbed8314a","_number":941744,"virtual_id_number":941744,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-15 06:42:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 22:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-15 00:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd6c508bc46b0856918d7feeb44489395d070edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 22:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75dc0e49d897ad27651594a10f3c1ed2d5eba8de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 22:53:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ef977cdbba44f8b40425cc091ee728\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2d61143dc7b432c8c2770d10a2ce52d : NODE_FAILURE Node(set) request 300-0026343527 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d6a6e2eeece78ca56f8ed7e2c16de380dc29f511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 00:00:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/beac6141f2664d63a299fa6adf3bb8e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e8929a7f4343d38081034251ce28eb : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2319e24257894b039aaf6dae8e1b874d : SUCCESS in 38m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f49a48de981946f4bce2f44997936668 : SUCCESS in 49m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a760c9173124f548156b4f71391dd9a : SUCCESS in 38m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbcc5399f67d45039580e38e783ceb5c : SUCCESS in 40m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a846e8f81894786985d92944e2048d4 : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dd52127fe064eed8e6b80a4a470653d : SUCCESS in 12m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3c85706dcd24ad284ba88eb327a09f9 : SUCCESS in 24m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94f6494bb5c24b229aa046a565909df1 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34a0e4ec67914e708ee5010d9279d881 : SUCCESS in 9m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab00298d10549b4897a60c522c3aefe : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/edf10d37b6924ccb87416dfeff0cc399 : SUCCESS in 22m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7546748887ff459db51d25d8030984c9 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1036e03eb99f48c9ab42ad5ab43c8510 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c91c416d6031480fb1ec2af285cf079e : SUCCESS in 9m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2aef1ad9a163483eaad71943bcdc318b : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8e105d77f97440b8a54da4b445394a9c : FAILURE in 14m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff68800d796442c1adb9b0f186989045 : FAILURE in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5695bc70e84a49339a56435834e9f6a2 : SUCCESS in 1h 05m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bda2087d0b714d37a0dc983188fb84ed : SUCCESS in 37m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87770908e5c74d779501edefe1f1653b : SUCCESS in 30m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a5ae2ffb0b84c2eadb5bea2233395d8 : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd20b2990fd4869bcb7cea4e4f32814 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2e83f286eb547f8be5085c123f61160 : SUCCESS in 19m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d6d6092d98d4dcc8d59fea0139e29d6 : SUCCESS in 46m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4501f3d6e56b4e2fa11478cc9c14dc4b : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"02c760c574c275b72902ac206fb52238da5879b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-15 05:26:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"54951db939fd1a0b9b3a1e50c4a388cf29933bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 05:27:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa05230c022c463c8043346583242566\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87cf96ecfb964e52af65e00f15d05415 : NODE_FAILURE Node(set) request 300-0026345576 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"660bf828d00e144eb2f84370379b5bc74e802d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 06:42:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb2a4b92264f5282d782c49065ac1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb740dcdb7af49a896fffbd4ee7602ad : SUCCESS in 5m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/389169bcd3eb42a6bc3b82a68ac01b8f : SUCCESS in 35m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/955600683f04497cadbad8152d717a7f : SUCCESS in 1h 08m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92ccb4edc3ae4648aa316add34a924be : SUCCESS in 37m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bfaf18676de4c1dba8161bc25ab3147 : SUCCESS in 44m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a08fa3d2051d4992812c79dc4082832a : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c1f401a8049485a84fdfd2638e3b15e : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9844a9fd64544bbb8412c200cc877914 : SUCCESS in 15m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5ffb8d1a9ed44248827fa94e2dffb43 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce7d05b4aef493dbcef5a72606e1097 : FAILURE in 9m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0370163aa7e40bea8198995c0d8807f : FAILURE in 14m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ebf83e1bfa354e4daf163c5b3c523fb0 : FAILURE in 18m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0db68e3584d43729d332e3f4b6ac318 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b3caa2497d547b7aca5902c5e72c6a4 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0db59a34fe1c4fdc9b76162803a5c0b6 : SUCCESS in 9m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1261672bc114b55b130c3372b89c608 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d9e30aaf9a04e7f82fd2de0146395c6 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e46cae0a733740ec8f65c5c9ea1a8fa9 : FAILURE in 7m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ad2edc140d34a8da1a4f39700d811ec : SUCCESS in 37m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dcfdd16c989c481eba472d98e731a83c : SUCCESS in 54m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/484691654d4842ed9099c179d28ab717 : SUCCESS in 33m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ea2fe37ae6140878bfd5b12960ae480 : FAILURE in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe8b0bdabfc435eb7c298a92432f6a0 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce3307b35c1c4ba98cb901e16a1aa5d5 : SUCCESS in 20m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfb828aef3ae4378ab2264a1a5454d6c : SUCCESS in 39m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ced48d210ce4cf5831bec795b76a9c5 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d9514e32f29eb3c8dd27f5a4a1a3492fbed8314a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-18 18:38:55.000000000","message":"Abandoned\n\nSquashed into aws-chunked patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"698943e4c1e520de0c0de464b50077bc7f0a7a26","revisions":{"5ece93d6719b2dd66f157909f1839afa9ef26788":{"kind":"REWORK","_number":1,"created":"2025-02-14 22:53:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/941744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/941744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/941744/1"}}},"commit":{"parents":[{"commit":"adc6d0acb708cf4e7567080c20a777092e68ba87","subject":"fledgling boto3 tests in functional/s3api/test_obj.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc6d0acb708cf4e7567080c20a777092e68ba87"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 22:51:56.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 22:51:56.000000000","tz":-480},"subject":"fixup! fledgling boto3 tests in functional/s3api/test_obj.py","message":"fixup! fledgling boto3 tests in functional/s3api/test_obj.py\n\nChange-Id: Ibe62d567caa4f7c495a222e539090db9aff0cbeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ece93d6719b2dd66f157909f1839afa9ef26788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ece93d6719b2dd66f157909f1839afa9ef26788"}]},"branch":"refs/heads/master"},"698943e4c1e520de0c0de464b50077bc7f0a7a26":{"kind":"REWORK","_number":2,"created":"2025-02-15 05:26:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/941744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/941744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/941744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/941744/2"}}},"commit":{"parents":[{"commit":"adc6d0acb708cf4e7567080c20a777092e68ba87","subject":"fledgling boto3 tests in functional/s3api/test_obj.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc6d0acb708cf4e7567080c20a777092e68ba87"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 22:51:56.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-15 05:25:55.000000000","tz":-480},"subject":"fixup! fledgling boto3 tests in functional/s3api/test_obj.py","message":"fixup! fledgling boto3 tests in functional/s3api/test_obj.py\n\nChange-Id: Ibe62d567caa4f7c495a222e539090db9aff0cbeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/698943e4c1e520de0c0de464b50077bc7f0a7a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/698943e4c1e520de0c0de464b50077bc7f0a7a26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
