)]}'
{"id":"openstack%2Fswift~941994","triplet_id":"openstack%2Fswift~master~I312258bf1089f35fa5736a20fd88768cb94b762e","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-17 19:40:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I312258bf1089f35fa5736a20fd88768cb94b762e","subject":"Add test/s3api/test_object.py","status":"NEW","created":"2025-02-17 18:09:25.000000000","updated":"2025-05-14 17:11:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5de7bff884157db6db44b8dd3bcfea88272db89d","_number":941994,"virtual_id_number":941994,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-14 14:50:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 19:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-17 19:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4f21e975bd277244169e76d6469520764084e14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-17 18:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29cea6db2d86926dd931496b01a35bfaf2f194da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 19:27:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cacfb6f32a84957b25cbe3e83be363c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85a1133942aa4f1bbcd48d5655714eb0 : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ae938c62379100f3fffe6c934a2eb2c57e4f1eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 19:40:49.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f48bd32a9be441c92bb8029f2a4d2de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e40d0b233ea44f7bb7c8e8e1dde6867 : SUCCESS in 11m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e5bca53a3924a11a6909681f1255200 : SUCCESS in 58m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e74b9fefd8d842aabd38a9aa4081dc65 : FAILURE in 20m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cb3fd9ef2eb442dbb55b3fff6d37d45 : FAILURE in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b080e36efa7043fea97216451feb5bec : FAILURE in 17m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58a4c81e04674394ae79032c0566d284 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ff16cbd47114b349e4d20904ff7f417 : FAILURE in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e70f587e8a74a0799e2a35f5e09aa2f : FAILURE in 14m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/357d4149c21b4313bf4e5c51d18c7a66 : FAILURE in 16m 33s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f19381fa77994a6b929547723b769a43 : SUCCESS in 7m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/964efcbd958e4da08cbfc3aa27c08253 : FAILURE in 23m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bde59770b8cc4069a744f6464c76f419 : FAILURE in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f3c17ee2954da398ea4d4e3e997108 : FAILURE in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aeeffcde0d304f319800187a1a77ae9e : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9aa7c409a57c233776016c69fa9b66522e797b6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-18 14:53:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c2515c6d2b304bfffe96169cbbdf363b22c7566f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 16:09:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e98836c51944678b6279402afacf782\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee8b33b7b9444c0bba3d39b42e4011a1 : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7ee6437aa17db4ae021ed1359734505c3b0df6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 16:11:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e6291fc83c948a8b810ab58b9b62e4f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d60674890294278aa5183d8e319a99b : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5404475d0d48dcb7c5fd2997240491 : SUCCESS in 9m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/53e2ed2285a04e668b82e8c74ed1c90a : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16d8add7b1e7467c870076e95465cef8 : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/feb53a9fa47b44288f7e3ca00e6928b1 : SUCCESS in 1h 10m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1ce58995ff74386b21cf491168e4893 : SUCCESS in 48m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05511456448e400eaea86e3ad07ff670 : SUCCESS in 48m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/633a931cee114068bad84427e252417d : SUCCESS in 53m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e71dbe77db84b08bbe4d91ce399ecd4 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df7fd70e0a6b435e807902bb622b4248 : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/711fa200df41437dbf3d43c3f2a71d2b : SUCCESS in 22m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b722f6ccb847474dab263fe193445e78 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35ccbddbe60544c7b6343141530c5704 : SUCCESS in 21m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3042566f213743ca973faf6354a68209 : SUCCESS in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e6597cdc903a4e3c9150a77375f7ea35 : SUCCESS in 31m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d7c75716c40548669b3b90022f08376c : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ab606e9d7cf4777819f4cfb3734b07a : SUCCESS in 17m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fe81a6b5fd84ee3879122e306acfd42 : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/708dad778f2847ff8cd3733863d485ba : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c79d9a30ebfa4e6aab0480c45010f77c : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62a5d1d8123948fa94f493d182d1471a : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0322733a1b749c699c99d4e572b5ed1 : SUCCESS in 55m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b10c1bd1e3a24bae8cf91b52c3b718f3 : SUCCESS in 1h 01m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20f21820a9ec46ee8b6ee363d75e63d7 : SUCCESS in 44m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64e0e61244e541fda11722d6addd5afb : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b395c8290d474a3bb373f8c359892bb3 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17481f5167dc4a59bd2105d00537d1a2 : SUCCESS in 18m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e6a8b65ac334b988c3f1a8fb82107cf : SUCCESS in 46m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71178c9d9ee74fb09e1884d53edb5c40 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":2},{"id":"cc741b3fc03be7a9ddccc15fb7a0e1372e29d99d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 13:25:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c7977ca98afe93fa2ec2a0f13da63fbaba871203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 14:50:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779f1be97f184a68837b4f400febfc72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9bc28abcced439485cb8d78e562a710 : SUCCESS in 10m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bed35ba3059f4bc99d96858be8b0ffd3 : SUCCESS in 1h 07m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e22951e856ac4bb18434b3759f74db62 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d889a81aaa14692996cf86b79598a10 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf8e8bc058594de496ed726ef272caea : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/305c08bdf57a41b9b4ea6062da0f2db4 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a6c810933b94a5e942687aa2ef30c9e : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/83ceeb2567a3421a9701725646cdf40c : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/89a766d22299490187891d896402c75f : SUCCESS in 13m 52s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7967e9383d4540cc8636f486e1d48b4e : SUCCESS in 10m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9356bdb321184fdfb891567ceacc66bd : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63a17f926296475390efd7cc075dad31 : SUCCESS in 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c425f12b39e40318394599c7d221d86 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/016a878f74cc441793926ddc1ab6c988 : SUCCESS in 21m 50s","accounts_in_message":[],"_revision_number":3},{"id":"5de7bff884157db6db44b8dd3bcfea88272db89d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:11:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c423791c434f0baf9c9aa1d6093f66\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f97931a1735346548153f901046da927 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b46bab7a8d9fbdcd1db1cef90d360e94b0125ee","revisions":{"039f356283d31d641a20a632c52c2100e3414994":{"kind":"REWORK","_number":1,"created":"2025-02-17 18:09:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/941994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/941994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/941994/1"}}},"commit":{"parents":[{"commit":"4185af65fac33a5ba42b2a46b40ac281646bd66d","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4185af65fac33a5ba42b2a46b40ac281646bd66d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 13:54:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 18:08:57.000000000","tz":0},"subject":"fixups for s3 chunked xfer boto3 functest","message":"fixups for s3 chunked xfer boto3 functest\n\nChange-Id: I312258bf1089f35fa5736a20fd88768cb94b762e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/039f356283d31d641a20a632c52c2100e3414994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/039f356283d31d641a20a632c52c2100e3414994"}]},"branch":"refs/heads/master"},"f93e03546ef1e582f60602257350a011f95e1fd9":{"kind":"REWORK","_number":2,"created":"2025-02-18 14:53:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/941994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/941994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/941994/2"}}},"commit":{"parents":[{"commit":"4185af65fac33a5ba42b2a46b40ac281646bd66d","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4185af65fac33a5ba42b2a46b40ac281646bd66d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 13:54:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-18 14:51:09.000000000","tz":0},"subject":"Add test/s3api/test_object.py","message":"Add test/s3api/test_object.py\n\nWhere appropriate, it makes sense to have s3api functional tests that\ncan be run against both swift and AWS S3.  This patch therefore\nrelocates boto3 tests from test/functional/s3api/test_object.py to\ntest/s3api/test_object.py\n\nTests are also added for put_object with more checksum types, and\ncopy_object.\n\nRequires the aws Common Runtime (crt) to be installed to provide crc32\nand crc64nvme checksum capability.\n\nChange-Id: I312258bf1089f35fa5736a20fd88768cb94b762e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f93e03546ef1e582f60602257350a011f95e1fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f93e03546ef1e582f60602257350a011f95e1fd9"}]},"branch":"refs/heads/master"},"2b46bab7a8d9fbdcd1db1cef90d360e94b0125ee":{"kind":"REWORK","_number":3,"created":"2025-05-14 13:25:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/941994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/941994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/941994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/941994/3"}}},"commit":{"parents":[{"commit":"27fbe95221dd55fc979a41b290455db2fa69c54c","subject":"s3api: Add support for crc64nvme checksum calculation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27fbe95221dd55fc979a41b290455db2fa69c54c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 13:54:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 13:24:54.000000000","tz":60},"subject":"Add test/s3api/test_object.py","message":"Add test/s3api/test_object.py\n\nWhere appropriate, it makes sense to have s3api functional tests that\ncan be run against both swift and AWS S3.  This patch therefore\nrelocates boto3 tests from test/functional/s3api/test_object.py to\ntest/s3api/test_object.py\n\nTests are also added for put_object with more checksum types, and\ncopy_object.\n\nRequires the aws Common Runtime (crt) to be installed to provide crc32\nand crc64nvme checksum capability.\n\nChange-Id: I312258bf1089f35fa5736a20fd88768cb94b762e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b46bab7a8d9fbdcd1db1cef90d360e94b0125ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b46bab7a8d9fbdcd1db1cef90d360e94b0125ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
