)]}'
{"id":"openstack%2Fswift~941995","triplet_id":"openstack%2Fswift~master~Ic1f915e6ababd91f7c48c6a644dac8129ea498d8","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-18 14:57:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic1f915e6ababd91f7c48c6a644dac8129ea498d8","subject":"WIP more s3 checksum func tests","status":"ABANDONED","created":"2025-02-17 18:09:25.000000000","updated":"2025-02-18 14:57:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a10b7327619347abbc2c89b3589f844c940e6aa4","_number":941995,"virtual_id_number":941995,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-17 19:50:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 19:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-17 19:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a84e6c14f171790068ed4453f72d9bc6d4d8f174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-17 18:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afff45102960fb30d8809bcb24c6e16468d41a05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 19:25:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/596a0f0a6a1f4d00901e58d44c0a0329\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7370d2c924c24436960d0bbc59f07b47 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"96f1f34b6de7b5a689886e4cbc9856e75e3a66bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 19:50:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/273d7fe2fb0d494eb256467e2c98f6bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6fd3c9e5ba44a79332da2e55dc728e : SUCCESS in 5m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0a444730ab14b34938b18935bd84776 : SUCCESS in 1h 03m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca1951c433b64926ad0daa9a1734859e : FAILURE in 24m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1425adc031f24bbbbd8415a6da8cde03 : FAILURE in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c65d800117a048098bd988fd660d3e21 : FAILURE in 20m 57s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a718c026953d4e56bcaa66fd8c8e9845 : FAILURE in 16m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8d0b1d49956405693139bc297d511e7 : FAILURE in 18m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc44c307a9944bec9cc11a997cc7b033 : FAILURE in 15m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b96fea2c695f47d28ca23bf812e8db1d : FAILURE in 36m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9395cb481e940978fc4ac57ee196866 : FAILURE in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba03ad4fd4f4406294726cfc58b62ab4 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7888a2db1f646b995382c74f9ddac8f : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a10b7327619347abbc2c89b3589f844c940e6aa4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-18 14:57:52.000000000","message":"Abandoned\n\nnot required","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bab23e607a6c2e71ffe46a7d3b1c860b07d9229","revisions":{"5bab23e607a6c2e71ffe46a7d3b1c860b07d9229":{"kind":"REWORK","_number":1,"created":"2025-02-17 18:09:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/95/941995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/941995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/941995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/941995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/941995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/941995/1"}}},"commit":{"parents":[{"commit":"039f356283d31d641a20a632c52c2100e3414994","subject":"fixups for s3 chunked xfer boto3 functest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/039f356283d31d641a20a632c52c2100e3414994"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 17:58:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-17 18:09:04.000000000","tz":0},"subject":"WIP more s3 checksum func tests","message":"WIP more s3 checksum func tests\n\nRequires awscrt to be installed\n\nChange-Id: Ic1f915e6ababd91f7c48c6a644dac8129ea498d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bab23e607a6c2e71ffe46a7d3b1c860b07d9229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bab23e607a6c2e71ffe46a7d3b1c860b07d9229"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
