)]}'
{"id":"openstack%2Fswift~942126","triplet_id":"openstack%2Fswift~master~I43096b1387cb85e4b9c9651b8b9aaedbb20dfba9","project":"openstack/swift","branch":"master","topic":"aws-chunked","hashtags":[],"change_id":"I43096b1387cb85e4b9c9651b8b9aaedbb20dfba9","subject":"sq? s3api unit test cleanup","status":"ABANDONED","created":"2025-02-18 16:55:42.000000000","updated":"2025-02-26 18:03:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9724137c696a716fe84f0e49828209c1e4c8982","_number":942126,"virtual_id_number":942126,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-18 17:52:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 17:52:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eeae44dffab87ecfe448889cf740ffd8d99d48a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-18 16:55:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecbd0000adc0a7a1290936a0c2948dc7a3ba2937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 17:52:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e806669726d40faa129404905e0a3e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eadfe5b547d4f3b93717311e684d423 : SUCCESS in 8m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6532af2a89584667afaf7b258936ea64 : SUCCESS in 50m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b59f6dfa9ae94a4abca4724d7ea2e417 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4839f9d47812446e84306e00dbfa523f : SUCCESS in 27m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a580bc803fe45259cb9172bb733ec16 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a5408d035c14b799ed39f4aa325b33d : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0ad98feaf3547fea3ef98c7696b9a6e : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7198ff5e2094dd58028cf5082e2a95b : SUCCESS in 16m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/53ff044581454a7e9f2e0a8c2bcc5730 : SUCCESS in 32m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c4ab45f2a3ec4d69a6aada2cfd3b1a72 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/339d6f845e3b40558aa0cb4d194feb9d : SUCCESS in 13m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/83b1b3bb4715407da6ed3e6678886cfd : SUCCESS in 14m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac773a0bbcc94eb79ea8372f26c4fa4f : SUCCESS in 38m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b867acbc2a049678e64a48ba050dfc0 : SUCCESS in 52m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06d5370e20684735aea18b0fa4abb01e : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa943bc2d6d431894e3cd317320fe88 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2876519f52c94bf1a68fb67ef169de8c : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ebebb68b014cbbc863aff7b645f21c293c2f4d53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 18:32:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587950a3091846b9a5f7a817589a5c68\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5acc8f82ca70495a852e57551da44fb2 : SUCCESS in 1h 33m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d9724137c696a716fe84f0e49828209c1e4c8982","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:03:48.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/836755","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49bcdab1a8685c44ebd0cf50e2a7a02c382decf9","revisions":{"49bcdab1a8685c44ebd0cf50e2a7a02c382decf9":{"kind":"REWORK","_number":1,"created":"2025-02-18 16:55:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/942126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/942126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/942126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/942126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/942126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/942126/1"}}},"commit":{"parents":[{"commit":"f93e03546ef1e582f60602257350a011f95e1fd9","subject":"Add test/s3api/test_object.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f93e03546ef1e582f60602257350a011f95e1fd9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-18 15:53:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-18 15:53:39.000000000","tz":0},"subject":"sq? s3api unit test cleanup","message":"sq? s3api unit test cleanup\n\nCleanup for the new sigv2 tests\n\nChange-Id: I43096b1387cb85e4b9c9651b8b9aaedbb20dfba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49bcdab1a8685c44ebd0cf50e2a7a02c382decf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49bcdab1a8685c44ebd0cf50e2a7a02c382decf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
