)]}'
{"id":"openstack%2Fswift~942309","triplet_id":"openstack%2Fswift~master~I4113e2fd47e5535eec8bd9487884af077e8b0318","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-20 22:58:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4113e2fd47e5535eec8bd9487884af077e8b0318","subject":"tests: Allow more configuration for S3 cross-compat tests","status":"MERGED","created":"2025-02-20 00:05:43.000000000","updated":"2025-02-20 22:59:40.000000000","submitted":"2025-02-20 22:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"942309","meta_rev_id":"85806ef08ec46427392fe41ae074211eab0c023e","_number":942309,"virtual_id_number":942309,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 22:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-02-20 21:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-02-20 21:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 00:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 21:42:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"a408984290847e121612cbaf6f453d10bb351540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 00:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"520f78660a8fbc7a81605f465f23cef1eade58f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ccce7dedf81481e865991e5a91b918b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6898d1ec7c614848a96b59d605c216d3 : SUCCESS in 8m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/026ba31f039047c6a65e5ac5209d8e5b : SUCCESS in 35m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a130a32e1a042e7a0e63ed37d915536 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d237cb799194b0cbbe7788a4f779434 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1aa21ea9b4d4e0d8a640d987c20ad76 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ab631806ac960a38d97b228368c419d45b5ab06a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 03:12:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6066a232bf1545e6829fa150cb81f82b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a962c243e0c84cd391ff16702e27c23c : TIMED_OUT in 3h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"04980f320c6b11d9227dc48a45e85938f790f445","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-20 21:42:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d969f8dd51421ef6b4bb4fb96a6fc4d61c478ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4107a7b61851ab08e43ffe14033ce15374bdec96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 22:58:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0743026a8acf49dab0bb2c8fd2be35b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe5588abe6147c3a281ba4edd871fca : SUCCESS in 8m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/12e902cf438e4b079b6b81dc21723f86 : SUCCESS in 1h 06m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16a92533bb4b4c58a794c73b0e2a57de : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/564b11c7a0734782ac2108063db35e73 : SUCCESS in 19m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cba9b37f82594390b1fbc8dc674f6cfb : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dc466f17ef4a4e6e9a49298bbe42e58b : SUCCESS in 30m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/427bc9d9d2044f7e86b98716a3313713 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6cfca3c7339f48b496c6a0caa5641b92 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d57d2db21fc149ea8e482c13718d8182 : SUCCESS in 20m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18079f638488469d82cb532ae7086e12 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2c9273f749f8458182a5a3c1c008ad56 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f52e4f2408048958c95e5dcea68b634 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6a45a267893fdb5be37dfabe2ca60ff97bba422e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 22:58:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85806ef08ec46427392fe41ae074211eab0c023e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 22:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f0feef308f45de9670245cc6943977\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843ff83f1c11417c99be4ed9592ba345 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1ae27244d57546db812751db02a2c839 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9354c9eb6752e583bdcdf7e8af658e262fb77b0","revisions":{"f9354c9eb6752e583bdcdf7e8af658e262fb77b0":{"kind":"REWORK","_number":1,"created":"2025-02-20 00:05:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/942309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/942309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/942309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/942309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/942309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/942309/1"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 00:03:57.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 00:04:00.000000000","tz":-480},"subject":"tests: Allow more configuration for S3 cross-compat tests","message":"tests: Allow more configuration for S3 cross-compat tests\n\nSpecifically, allow endpoint and region to be configured. This allows\ndevelopers to compare boto3 behaviors with TLS enabled/disabled, for\nexample, or AWS behaviors between different regions.\n\nChange-Id: I4113e2fd47e5535eec8bd9487884af077e8b0318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9354c9eb6752e583bdcdf7e8af658e262fb77b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9354c9eb6752e583bdcdf7e8af658e262fb77b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
