)]}'
{"id":"openstack%2Fswift~942310","triplet_id":"openstack%2Fswift~master~I02efba463a8263ca44be511c025c6c6bfbc57334","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-20 11:32:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02efba463a8263ca44be511c025c6c6bfbc57334","subject":"tests: Simplify test bucket name","status":"MERGED","created":"2025-02-20 00:11:28.000000000","updated":"2025-02-20 11:33:56.000000000","submitted":"2025-02-20 11:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942310","meta_rev_id":"9cef8ec1308c2f4d89de139d071b6581fde66a76","_number":942310,"virtual_id_number":942310,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 11:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-02-20 10:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-02-20 10:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 00:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 10:39:21.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"7dc88849da2da9bbfa5680602c5ccef2afe7f9ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 00:11:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a943598d25393bf9b88a3f9e0a38156d1a6872e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb0955fc93742158fa1098651a0d3a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661efce22eda40b8aef865d05755d8fc : SUCCESS in 6m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62acaad013524208a32fc1c6a6b5dfdf : SUCCESS in 46m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/218a557883fe402092e0cd5a334cdb43 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37a0920066614ba1bbc08120ddf23f55 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc98bcc524bd41f38cd1ed4906837e77 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8518a3b4e5f5efb3c3f05f58fa3fb33cf192b013","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 03:16:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22f3638d854a4fc1b263e105768bd8d5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b7250b5c56c4cfb9e0f21f97cfb91cf : TIMED_OUT in 3h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"64156b9e9660aeb99ed57f5ed9d44de819b57934","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-20 10:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eec6a02fb09aa71e9190a20a4efc4450bc3c770d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:39:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce1cc573d5b02f439e2b66c58913c6a38bf0a37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:32:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18fae1800084717afcf878cf44d65fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcd08327e9e41a0ac8bbb87d0ab02a1 : SUCCESS in 9m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6fe68a789b6343b7b4ddcdd578174f68 : SUCCESS in 49m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ffccc61f00764271831f8c304e3365bc : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea917df6b854795864291c41ecddd35 : SUCCESS in 23m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3552c6feee324d7f920b24d208480347 : SUCCESS in 17m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b3026e487bc4d71850d1f9dd97cf833 : SUCCESS in 22m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cc56a7756c814af0a237a8d6720fb746 : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/645baaa36be44a2aa26068f40176d602 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1d434fac26dd468c99157de1bfef4379 : SUCCESS in 14m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db0243244ddd458292909c948059a589 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a4d0e5247bb4dca87013557ca8eb467 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb944170648b4059bf4f747cb09950df : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e54027b19c9c25b45182ace36b0c5f08192a50c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:32:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9cef8ec1308c2f4d89de139d071b6581fde66a76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc3dbd8321c400e9eea1a2b7ac37646\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8c41e555cc4568b1c9308bf65b7131 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dd242697deeb4dfb89426ea4f77fdb8a : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af","revisions":{"ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af":{"kind":"REWORK","_number":1,"created":"2025-02-20 00:11:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/942310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/942310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/942310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/942310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/942310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/942310/1"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 00:10:12.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 00:10:15.000000000","tz":-480},"subject":"tests: Simplify test bucket name","message":"tests: Simplify test bucket name\n\nWe already prefix everything with \"s3api-test-\"; there\u0027s no reason to\ndouble up the \"test-\" part.\n\nChange-Id: I02efba463a8263ca44be511c025c6c6bfbc57334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
