)]}'
{"id":"openstack%2Fswift~942329","triplet_id":"openstack%2Fswift~master~I4e13a502eae25bb61d8f0129f106f7e0ab4896e5","project":"openstack/swift","branch":"master","topic":"alt-s3api-version-fix-patch1","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-05-30 20:47:57.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-30 20:47:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4e13a502eae25bb61d8f0129f106f7e0ab4896e5","subject":"tests: Fix versioning cross-compat tests","status":"ABANDONED","created":"2025-02-20 01:06:15.000000000","updated":"2025-05-30 20:47:57.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"66c92eab67299ab22cc4043e4ffb114f9203a306","_number":942329,"virtual_id_number":942329,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-20 17:47:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-05-13 14:59:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-02-20 16:55:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 02:08:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 16:55:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-02-21 12:24:11.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-05-13 14:59:25.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f30c8edad3491dddea9f93b549077cd478de31a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 01:06:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2604a072348493729cccc729c1445f829ab046ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 02:08:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70aee550431543d98b53f964ca1ae7e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68588b5d90224930acf419e4853f1876 : SUCCESS in 8m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9eec1d3be7bb46e29ad29140e4a216f5 : SUCCESS in 54m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ab37198b5ab49e38d5e46dacbe4d6b7 : FAILURE in 5m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9668681da86477c86a5784308a7f981 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4e97a0c68d49139357b2e4e5a093e7 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0f4da29efb9e86c8a390b9f2f4b17140b02300f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 04:09:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd32e3c871734ae8a3fdc01b300a17e0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20d93eeecb424e7c81dbed9bcfaeff80 : SUCCESS in 2h 50m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9621a6e335681662ef74e1dac73ac18b834f5045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-20 16:46:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"24dc92d41abd522b9aefa57e4a2f10a961362fc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-20 16:48:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"78cf5de1272c0fc67cb3905772d0823c1e47013b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-20 16:55:08.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f01a47a6bd5dca0f0c296e182d6134c8a8c8e917","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 17:04:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12f5356e2ad7d44811d280572f892b582489f20a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:47:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009c60f610cf4ccf90208bd2b8b0a75b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89595b2ef5f348ab96598c67bf988ef8 : SUCCESS in 5m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23b8ed7e38984990ae542291201a9dd3 : SUCCESS in 56m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c5a23d06d3f425bbbeb325c98258ae5 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f41f79a419848fdb2aab635d5a3007d : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db705e998af4627952f92f30f5852dd : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c03d4cdcc2455e81fbd8f40815cfe85703004130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:49:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d82087a39c45319c4e8f0b58721792\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07d687f075d545cca1799d2b326444b7 : SUCCESS in 2h 46m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0260b2130a6b08c623a8252776902c267f5e2fff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-21 12:24:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9147d8e9c750bb75ed5d65892135ce5e13ea275a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-03 21:05:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aae10998b2bdbb745f5c145242e9539ca7d6c967","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-13 14:59:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66c92eab67299ab22cc4043e4ffb114f9203a306","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 20:47:57.000000000","message":"Abandoned\n\nDone better in https://review.opendev.org/c/openstack/swift/+/951443","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25407931bd5cf49df32925c2c26d32240362aee2","revisions":{"4d3c49d42bf15b2a517a42d886a3597bb47186fd":{"kind":"REWORK","_number":1,"created":"2025-02-20 01:06:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/29/942329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/942329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/942329/1"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 01:04:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 01:05:02.000000000","tz":-480},"subject":"tests: Fix versioning cross-compat tests","message":"tests: Fix versioning cross-compat tests\n\nThey don\u0027t pass against AWS currently, since newer boto3 will opt into\nadditional checksums (specifically CRC32).\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/940172\nChange-Id: I4e13a502eae25bb61d8f0129f106f7e0ab4896e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d3c49d42bf15b2a517a42d886a3597bb47186fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d3c49d42bf15b2a517a42d886a3597bb47186fd"}]},"branch":"refs/heads/master"},"2befce3d7dbc5f4b342776b3622697fb014f5d53":{"kind":"REWORK","_number":2,"created":"2025-02-20 16:46:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/942329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/942329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/942329/2"}}},"commit":{"parents":[{"commit":"ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af","subject":"tests: Simplify test bucket name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0b68e1b724d76019ac5a7e4e6d9bb2118cf8af"}]},{"commit":"b49941c9b872c22fdb4baf8c4922f1cf1ed13131","subject":"s3api: Simplify HashingInput signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b49941c9b872c22fdb4baf8c4922f1cf1ed13131"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-20 16:45:48.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-20 16:45:48.000000000","tz":-360},"subject":"tests: Fix versioning cross-compat tests","message":"tests: Fix versioning cross-compat tests\n\nThey don\u0027t pass against AWS currently, since newer boto3 will opt into\nadditional checksums (specifically CRC32).\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/940172\nChange-Id: I4e13a502eae25bb61d8f0129f106f7e0ab4896e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2befce3d7dbc5f4b342776b3622697fb014f5d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2befce3d7dbc5f4b342776b3622697fb014f5d53"}]},"branch":"refs/heads/master"},"25407931bd5cf49df32925c2c26d32240362aee2":{"kind":"REWORK","_number":3,"created":"2025-02-20 16:48:33.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/942329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/942329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/942329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/942329/3"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 01:04:58.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-02-20 16:48:25.000000000","tz":-360},"subject":"tests: Fix versioning cross-compat tests","message":"tests: Fix versioning cross-compat tests\n\nThey don\u0027t pass against AWS currently, since newer boto3 will opt into\nadditional checksums (specifically CRC32).\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/940172\nChange-Id: I4e13a502eae25bb61d8f0129f106f7e0ab4896e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25407931bd5cf49df32925c2c26d32240362aee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25407931bd5cf49df32925c2c26d32240362aee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
