)]}'
{"id":"openstack%2Fswift~942331","triplet_id":"openstack%2Fswift~master~Ia754021714a8ed3d5d6010a946c558552821fbe4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-20 17:22:23.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-20 18:28:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia754021714a8ed3d5d6010a946c558552821fbe4","subject":"tests: Remove a couple more py2 code paths","status":"MERGED","created":"2025-02-20 02:00:22.000000000","updated":"2025-02-20 18:31:07.000000000","submitted":"2025-02-20 18:28:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942331","meta_rev_id":"1e3c63f68e59a22a934e99bec6286d325fc26d77","_number":942331,"virtual_id_number":942331,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 18:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-02-20 17:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-02-20 17:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 03:59:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 10:20:16.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-02-20 17:22:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"429bd26c1a3e9734108e718a7f3cfe1e3bc6d09b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 02:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d19204dcae3ba38152d793702119a2d46598c754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 03:59:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90a0d1c3eda141e9b8b4415eaabc2f3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c6df7c47f44a4e8e84d6ad1c8caa6f : SUCCESS in 8m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7440c1ed74fd44f499ef4d67ada2a85b : SUCCESS in 1h 09m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e789ac91070b44a99997b59ed09e6cef : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc3ac6891f0d41658e0690cd10cc49df : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2220850970bb4b4eb4e9bf082c254b9d : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84770898325c4921a38dc5fbd8837ce3 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd0834238ef402c93427b085ef5fead : SUCCESS in 17m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32fa290d426e45f8aa9097257cd2e2f2 : SUCCESS in 10m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/82755242746c46c1836bccf712f53ef8 : SUCCESS in 22m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e6e15d5457994fc894c1593316cc9e40 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ceed3f767eb24becb37c92f8c378bc89 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/017082bddb4147e79e9cfe8c5529c0b6 : SUCCESS in 18m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37192d68a1c2491d8f4a3ea1c52a4154 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecbdb9ab9f6a40efa04cff30ef945c1c : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0695c5d2ada4ed6b1ed8902ee4e6ada : SUCCESS in 41m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4989e499c2646e79d80413a625c512b : SUCCESS in 34m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72b256fa860a41e19261fda190d05dc0 : SUCCESS in 52m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4172bac6fd4d40db96d0acc5d85fe000 : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4d086d2dc1400f9cb8a73b4f921478 : FAILURE in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90b90a26a1894855a139e8db8cc9e962 : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1906bc497a8cef2a84a4b271bf81fc0c8eaca782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 04:47:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a90e3bc5d074450976e0821661b13f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3de3a6dfaa874082a48eec8b38d4e8f4 : SUCCESS in 2h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c2e3c453afd5fd6d36f7954d39eb88c3a5682734","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-20 10:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05df25fa20e5ec731b0d5d5eae4d3cc20428cebd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 15:32:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c38cb73894151d13ab91db4137bedbbf80b7b279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 15:32:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"18b56ce7922e89330fae7efd1bac5c67556fd5e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:00:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/271ff991eb8344de9a7ba3d3059f94c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ddcdadd319b47e689938f0d4a70747b : SUCCESS in 8m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8f5d3993f9b4a5ca449e6357032ce37 : SUCCESS in 1h 19m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f039d31d65e4096822cbd94395fa7ca : SUCCESS in 12m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c2288fc8271455485d0d56433370b4c : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e876fa99e2df44adb928f71c050fbf05 : SUCCESS in 12m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d6fd0d3ba85496291ec4caa98a21499 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7880aa865a342c4b13d29d0e220069f : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a4b513cd6ac41b8af7bc22aa95d85a3 : SUCCESS in 10m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01dc6a760a194e778459e9c45efb5b11 : SUCCESS in 40m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d8a55877765427fb97cba0379b81f52 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4cf56fcb813a43bb8f6a22e51e0e4d24 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd9ba001b39e4e4ab11f268f7f9f0cb1 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/765f533eea40473f893a68180787915c : SUCCESS in 7m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f5b0ccf21a4464da827ef71174115d8 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7b2b01a3c9e4b20bbcdf9cc42c53efc : SUCCESS in 1h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4c39411b78f42698cb918ebcf174a56 : SUCCESS in 54m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/813ed7e42ec044e7a46ac48d59a6cb6c : SUCCESS in 52m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8dd33095e874672bde55f167bf1a433 : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6154463ab24f25a0c2fb1d3b0f289f : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f81466cf69d64576847920dac3c02b60 : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":2},{"id":"314175c457039c4a193453607a8f5f5fab807312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:03:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d4ed59f1a6418e9ecd312eaf2650cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eee0cbb30abe4e94aae2898a90c30334 : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9ed0ee470c9cb03fe2f97a3379dac7bc83d00fe5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-20 17:22:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab76fac1b70531dca378b593a2f4d961cc367df5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:22:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62a2945dc7e22e41d68cf2b37de0c68163513e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:28:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92cada32b244f5eaf32a120222657ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abf49ba9466448680db255f68edc890 : SUCCESS in 8m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d26b4f8ed81434d8add674b1333a243 : SUCCESS in 35m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b671ae28afea4cb3b1255e681df2db26 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/159826a9c59c4c8f8b6e5285b540066b : SUCCESS in 20m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5aa4519714ef469d97c44a1c961aa2e7 : SUCCESS in 21m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b101738979c442e8aa735373f1a1c920 : SUCCESS in 36m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ba9b7a6ca4d440eb9c1b27e7b8da3664 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c40963c3d5724613951990edb9804e37 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/519bc56a1e3f48c497e3a7118a3b4163 : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4673548c8354c09baf49af32704d65a : SUCCESS in 7m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b65fd811f53340c3916db2bede76c595 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9051ddbddccb47aab9a8401c5edd39b8 : SUCCESS in 55m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dba4da8789b54586bd9688d6cff9c388 : SUCCESS in 57m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3bbd27b2b81a4e4abd06be9afb67b876 : SUCCESS in 34m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/023fdc599dd74e4983b2715e3a790dd4 : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c5bb84aa024b3ea7da240b40db6132 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a94b1338c50b46a695ddcfe38cf6395d : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2610f24353fc6baf9405e8148241e8aed09d4f2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:28:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1e3c63f68e59a22a934e99bec6286d325fc26d77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:31:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b66a500e2b34909be3f498aed4202b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77860409de042a090fc25ae8d710bf0 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aeec1d0442e84a1792750653f9adef2e : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"605c70123594f1fb6141f3315e7399407457a282","revisions":{"d3a7e859166fa4398769a4c3f5de7eba4e5043e1":{"kind":"REWORK","_number":1,"created":"2025-02-20 02:00:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/942331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/942331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/942331/1"}}},"commit":{"parents":[{"commit":"747b2a4a328b62585444de37f96921b1401bf565","subject":"Merge \"Add per-container storage policy to account listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747b2a4a328b62585444de37f96921b1401bf565"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 02:00:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 02:00:00.000000000","tz":-480},"subject":"tests: Remove a couple more py2 code paths","message":"tests: Remove a couple more py2 code paths\n\nChange-Id: Ia754021714a8ed3d5d6010a946c558552821fbe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3a7e859166fa4398769a4c3f5de7eba4e5043e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3a7e859166fa4398769a4c3f5de7eba4e5043e1"}]},"branch":"refs/heads/master"},"605c70123594f1fb6141f3315e7399407457a282":{"kind":"REWORK","_number":2,"created":"2025-02-20 15:32:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/942331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/942331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/942331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/942331/2"}}},"commit":{"parents":[{"commit":"747b2a4a328b62585444de37f96921b1401bf565","subject":"Merge \"Add per-container storage policy to account listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747b2a4a328b62585444de37f96921b1401bf565"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 02:00:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 15:32:47.000000000","tz":-480},"subject":"tests: Remove a couple more py2 code paths","message":"tests: Remove a couple more py2 code paths\n\nChange-Id: Ia754021714a8ed3d5d6010a946c558552821fbe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/605c70123594f1fb6141f3315e7399407457a282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/605c70123594f1fb6141f3315e7399407457a282"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
