)]}'
{"id":"openstack%2Fswift~942350","triplet_id":"openstack%2Fswift~master~Ie59f11874daf9f166a699fa904581830942163a9","project":"openstack/swift","branch":"master","topic":"p-s3api-test-bucket","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-20 20:16:50.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-20 18:18:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ie59f11874daf9f166a699fa904581830942163a9","subject":"tests: More simplification of s3api test bucket naming","status":"MERGED","created":"2025-02-20 10:39:01.000000000","updated":"2025-02-20 21:21:04.000000000","submitted":"2025-02-20 20:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"942350-p-s3api-test-bucket","meta_rev_id":"fb8638d183da27c913e8b61bb6362c7ce2df3c0e","_number":942350,"virtual_id_number":942350,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 20:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-20 18:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-20 18:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 11:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 14:59:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"6a70f75afd43629ca992f7f1503f8a159f400bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-20 10:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9b71ee44dc9a050d21fc667f4a0404b71e71320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:29:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4c4033ee4b4d11b6e0a144363d4ba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10eaf18569654b69814d4633e55271ca : SUCCESS in 9m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ccf36c40d82a4cc29d384a38253949c1 : SUCCESS in 48m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cabba6af5191420cb1f4b845b561cf3b : SUCCESS in 6m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/77b2548023a54585b01ef80b171e7f26 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51e822b4fef4694a44022db5246a746 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"977fc8e8c613cf7c73b2094e2767c0a4f34f48f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:48:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4727a1223e8b415d8b48a06d5d731b78\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/03c31ee4214d4116be56fee27d102010 : SUCCESS in 2h 05m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f9e5ac541f7811f9ba8e88abbe6023159ca9f718","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 14:59:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9d63afaef18894fdaedc6b1ba096b04478f5d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-20 18:15:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f105d95d403b80341c910e6d6a44feb579f9181","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 18:18:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b39121ab8d626b256866e8862625c5984b17f556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:00:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f48e5ec46e84b4c8e1f1982138dd70b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62b55ab0d6f476cbb47cef582a9ae04 : SUCCESS in 10m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6588e451f7d4038ae5767c1439919d1 : SUCCESS in 42m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5d8a23b55634c51b13458d4a7fb821d : SUCCESS in 7m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84b51af5735044cfb114a8c3be25595b : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2053988dba4b22bec07ec9ec197905 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7b061bf5a3d15b7b203fc997ded04f674663e5a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:01:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b538e053669b1245d8994f60a8c9f067cf448cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 20:16:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca8344385b04ea5859a0971dc02e279\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df906f6fa0d42b1b48fd6223b74981c : SUCCESS in 12m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/261b0b7eb0c1436aa961834d35ad042e : SUCCESS in 58m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ef7eb71617504614b48e22168662a16d : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d06c3308c8364cb6b1a8ce59fbe26277 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53873f7007c24ccca9e9fb2d9500c716 : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/769c1737068d4b0894e1afdaebee0380 : SUCCESS in 22m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/10ff605ef5814f26a4fbcb44cecc52a0 : SUCCESS in 9m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/09937a14d33249748646a9bacca68b03 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8f05a748d2f0456da8f68a5d2b6e5f9b : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b01a97f29bc34caeac97343c4ff6ddb9 : SUCCESS in 4m 50s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/09a48a01e04f466dbf2f6717aac2b6a7 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681fa6fb4ab4465fafaf702b00093556 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8e030dc843d8831dafa4630e5cb9ca12a2acefd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 20:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2789e8e0e349eb9aea41a6ba00b7b9c4abc04d6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 20:20:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cb0cd1eb3d4e42a065ca1647abc929\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b717dd69193b4bd2b0b0c107e756f1ce : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d2e616c8bd8342b7a0b4656e793c79e7 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb8638d183da27c913e8b61bb6362c7ce2df3c0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:21:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcacf840afb8481c892c3a6b70b4e026\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf353a9153b44fc8ada4c618f4fdcf89 : TIMED_OUT in 3h 01m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"adc7760ed5639b202bef1905a463eb54185dbb80","revisions":{"34c20d597d2e8995c3d5c2f3b63f18a6362f97ca":{"kind":"REWORK","_number":1,"created":"2025-02-20 10:39:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/942350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/942350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/942350/1"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-20 10:38:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-20 10:38:08.000000000","tz":0},"subject":"tests: More simplification of s3api test bucket naming","message":"tests: More simplification of s3api test bucket naming\n\nChange-Id: Ie59f11874daf9f166a699fa904581830942163a9\nRelated-Change: I02efba463a8263ca44be511c025c6c6bfbc57334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34c20d597d2e8995c3d5c2f3b63f18a6362f97ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34c20d597d2e8995c3d5c2f3b63f18a6362f97ca"}]},"branch":"refs/heads/master"},"adc7760ed5639b202bef1905a463eb54185dbb80":{"kind":"REWORK","_number":2,"created":"2025-02-20 18:15:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/942350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/942350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/942350/2"}}},"commit":{"parents":[{"commit":"8b6abfb85d5986dec0333b03784f7a463c4cd5a6","subject":"Merge \"refactor test for x-delete-at w/t part_num and x-open-expired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b6abfb85d5986dec0333b03784f7a463c4cd5a6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-20 10:38:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-20 18:15:25.000000000","tz":0},"subject":"tests: More simplification of s3api test bucket naming","message":"tests: More simplification of s3api test bucket naming\n\nChange-Id: Ie59f11874daf9f166a699fa904581830942163a9\nRelated-Change: I02efba463a8263ca44be511c025c6c6bfbc57334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc7760ed5639b202bef1905a463eb54185dbb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adc7760ed5639b202bef1905a463eb54185dbb80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
