)]}'
{"id":"openstack%2Fswift~942464","triplet_id":"openstack%2Fswift~master~If65ca0e529c5e919e95cfbb6a43a6d11df409a9e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-24 11:54:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If65ca0e529c5e919e95cfbb6a43a6d11df409a9e","subject":"sq? refactor input wrappers","status":"ABANDONED","created":"2025-02-21 15:30:54.000000000","updated":"2025-02-24 11:54:32.000000000","total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"9a6fc91794b1af137481eac8eb2ed78d268e12e7","_number":942464,"virtual_id_number":942464,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-21 17:30:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 17:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 18:50:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"97636b6efcab01bc30c8bcbcffcf0f33f87ec8c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-21 15:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb5ae13d53e0175d5a13406efe98b18cbc8ff6b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:30:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de92be2a2a2c44aa8f9189dcf6f3c0dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b981726d80436e8c67a5b6d8f75c8a : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/dba3ad8ad78b490d96a152b05bc0921c : SUCCESS in 1h 00m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5aff4eaab63b4fff955585897527e23b : SUCCESS in 35m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60f19601926e44ca9a80ff808e131359 : SUCCESS in 1h 47m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66ae3c5c37654b59b5885c622fac3a08 : SUCCESS in 1h 08m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eaa01c37798840f3bea0d61a7a49d0d8 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4218e37504424adea546cd786ad59abc : FAILURE in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65028a48f92f4e508cf1d80fa9dfcd0f : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b4c29091aff46db96f131d97ba97b0e : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd78300032bd47d48fb15651111a2985 : SUCCESS in 22m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef25be4af08843d28eabba5c724d822b : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51998f943565469da155abdd804d7a42 : SUCCESS in 40m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0516a27ade0a485db1261b7c90182c5e : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46ebc784a49145848e56b567387c3aa5 : FAILURE in 15m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6dd77d79b2a14bd0a03ab19e49a65cf7 : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ad138e2885a4488b00ac961c4cf5420 : SUCCESS in 8m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf3345442646466489f222e2f726e6b3 : FAILURE in 14m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a92b9ae19e04d1ebd38e8011aac1cd7 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c317e102048423a9d77c107aea3d6e2 : SUCCESS in 51m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ab6377ddce24de09f9a7025c8c5e6f0 : SUCCESS in 46m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad216c6de4dd467fab895f8b8b7b4434 : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55b8e2de7a3c47f0b9d18fc17c6d9592 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef30868dea6a4bd5a4324d723bc157e3 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f9bea130d5241c9a830e0e77c7b6bb5 : SUCCESS in 17m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/747212b69e684043964b96d40880b746 : SUCCESS in 44m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79737b30961e4d0195a0fc0b59a1a4e4 : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":1},{"id":"100063411fb2ced7316d3a5e4c815b15119e1a68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 18:40:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f5af5726a14c24b981253dd0641050\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/195317c3dd614690bf0a232e3b0d662f : TIMED_OUT in 3h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"159d394a60c740bc9a6c7acc475cb25a5c97fe94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-21 18:50:43.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9a6fc91794b1af137481eac8eb2ed78d268e12e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-24 11:54:32.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a1ca4b6260ba23b3f0462f1c2037c464aeb57f1","revisions":{"0a1ca4b6260ba23b3f0462f1c2037c464aeb57f1":{"kind":"REWORK","_number":1,"created":"2025-02-21 15:30:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/942464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/942464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/942464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/942464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/942464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/942464/1"}}},"commit":{"parents":[{"commit":"66103a90deb743faa73d0bff2177f8494b05b1e6","subject":"s3api: Add support for additional checksums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66103a90deb743faa73d0bff2177f8494b05b1e6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-21 15:30:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-21 15:30:12.000000000","tz":0},"subject":"sq? refactor input wrappers","message":"sq? refactor input wrappers\n\nChange-Id: If65ca0e529c5e919e95cfbb6a43a6d11df409a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a1ca4b6260ba23b3f0462f1c2037c464aeb57f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a1ca4b6260ba23b3f0462f1c2037c464aeb57f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
