)]}'
{"id":"openstack%2Fswift~942522","triplet_id":"openstack%2Fswift~master~I6c46d854a9fce6242f63afd56d89bc620bb66427","project":"openstack/swift","branch":"master","topic":"fix-constant-s3-list-bucket-creation-date","attention_set":{},"removed_from_attention_set":{"37337":{"account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"last_update":"2025-05-06 18:23:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6c46d854a9fce6242f63afd56d89bc620bb66427","subject":"Fix constant creation date output in aws s3api list-buckets","status":"ABANDONED","created":"2025-02-23 03:20:46.000000000","updated":"2025-05-06 18:23:43.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b50d07bdf218080d179499293ef57d680eab1ad7","_number":942522,"virtual_id_number":942522,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-23 05:15:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-23 05:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-23 05:15:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a8dfb4d5eae0d3e0cae35d134b01e3fe477b7f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-23 03:20:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3197c95a8824d435b049059dd2474c73dae71f42","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-23 03:26:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc74561c9fb5ad464c43b07114780b9be551eb2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-23 03:52:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1787b048f28d64217f2e95ef556d2ded58f7d6da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-23 05:13:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27d1b1e3403f481ea7cb5c55379fc5e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0124d4c78dd24794bbd27f548002e7e5 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":2},{"id":"30ee75f6636062ae9c002023185076c87069812c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-23 05:15:21.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c95094c9def74a5986abd698539d93e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d813cb382334548a52a34d0e660068e : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/65efcc95f4684d4a8bc68a71043af0dd : SUCCESS in 1h 00m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed871c8108124b3f926a66af35a760ce : SUCCESS in 1h 15m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32db53efc07a408f8b73a829f7b46d64 : SUCCESS in 53m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ac2fedbe3e34a7ba639e859bee89f06 : SUCCESS in 44m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b30f5039fd44e4d9bb988f066893f01 : SUCCESS in 11m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cd4ee5beca049de88d3ec384a01a5bd : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83eea56314844a558c9e83a9894107b1 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/309529bb13724b8bb8fd29ca6c72146a : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e05c71444f274751acdec2ad4578e683 : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/614adfd72ea14d83851ee578196761e2 : SUCCESS in 24m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a43d837513754a9f9bcd41ecfb002d26 : SUCCESS in 18m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7f52136cb6de4f0bb60a9d88ea2a7a62 : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d4d02418eac415fb91c0797ac78d69c : SUCCESS in 17m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee50fb7dd81d4c40b41802921ccc90c5 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/daf0922d1405447ba75ecca71b65f8e2 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95a215e10f0f447f98ab52f95c82be63 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e5a99e07617455a9f9415935da69cc8 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57571a4e9e85419d8c91efdd5ccf3854 : SUCCESS in 53m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55cac3171f9b425388364aedc0ec0a3f : RETRY_LIMIT in 8m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1859124b4d334f8caf7ad5fe50ff11ce : RETRY_LIMIT in 7m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49a569d9f11c444493228950b13a5aab : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a84356f1244c5995f2db23a2634308 : FAILURE in 2m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05f70d4a913741d28d270cc3c53939c3 : SUCCESS in 17m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64d517783aee4d20922980228388c0eb : SUCCESS in 36m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7caf9245fe24e5886fb7ef23b48d136 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b50d07bdf218080d179499293ef57d680eab1ad7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-06 18:23:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1371317be6f560dc1f094eb7b366f7685a1a3179","revisions":{"b8eb603b42850eb1309cd8ad351937e11a84fde6":{"kind":"REWORK","_number":1,"created":"2025-02-23 03:20:46.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/22/942522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/942522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/942522/1"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-23 03:19:43.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-23 03:19:43.000000000","tz":-300},"subject":"Fix constant creation date output in aws s3api list-buckets","message":"Fix constant creation date output in aws s3api list-buckets\n\nChange-Id: I6c46d854a9fce6242f63afd56d89bc620bb66427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8eb603b42850eb1309cd8ad351937e11a84fde6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8eb603b42850eb1309cd8ad351937e11a84fde6"}]},"branch":"refs/heads/master"},"1371317be6f560dc1f094eb7b366f7685a1a3179":{"kind":"REWORK","_number":2,"created":"2025-02-23 03:52:27.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/22/942522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/942522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/942522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/942522/2"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-23 03:19:43.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-23 03:52:11.000000000","tz":-300},"subject":"Fix constant creation date output in aws s3api list-buckets","message":"Fix constant creation date output in aws s3api list-buckets\n\nChange-Id: I6c46d854a9fce6242f63afd56d89bc620bb66427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1371317be6f560dc1f094eb7b366f7685a1a3179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1371317be6f560dc1f094eb7b366f7685a1a3179"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
