)]}'
{"id":"openstack%2Fswift~942843","triplet_id":"openstack%2Fswift~master~I1fb53b4dc9ee38529662e0d4b5ae586990b162aa","project":"openstack/swift","branch":"master","topic":"p-systags-2","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-28 17:58:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1fb53b4dc9ee38529662e0d4b5ae586990b162aa","subject":"Refactor and add some encrypter unit tests","status":"NEW","created":"2025-02-26 18:28:17.000000000","updated":"2025-02-28 17:58:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f8990ae55154c1fdc2f70990ec19ba2813c975b","_number":942843,"virtual_id_number":942843,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-26 19:42:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2025-02-28 17:58:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 19:42:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"005474daffd19fdac6811d38957a2a6979fc83b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-26 18:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22bb0f067ee3c561286749324a5802883cfea5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 19:42:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef9f6ae90a2e4a798e953dde797e0d6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da527812e1c424e9c887f84811c9dad : SUCCESS in 8m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47451bcc99f9459a8f5907cc983a275b : SUCCESS in 56m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/363dd125631a41e58b870139c592f14c : SUCCESS in 12m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e10e4a020d4df4a031f18f4402f74c : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1b8d20486004358be95eab8a12ed9fd : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d97d39fefc64325853aeb1685242e58 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00aba5cfb6df4df9ad5b47c5080dab19 : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed40e3c7e69c4032ac9b0a9c40aa4168 : SUCCESS in 20m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c0a8e5b54aa4a478e30145164bcdb77 : SUCCESS in 26m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff497ff1ac1041c5bd3fb53714d192fd : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/be14b83bd80942239b283b6ae8a0462e : SUCCESS in 19m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e3cb94e981742ae8def853739700a73 : SUCCESS in 20m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd9500462d0c4cc3ba33d3a8b11f9993 : RETRY_LIMIT in 11m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b941413b247644e49e38179f8fa603b2 : RETRY_LIMIT in 11m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59bdf454361476f9fa38087a67bfb21 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16f15a873dd441fa7bd64e088c92ee6 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5dd45fde88484e69aa7d80705ed983f2 : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"868dd2ac456a91bf5875913f1f992632f9874f7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:18:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aecfd1fe32b48df8d65fd9f9c66ca4f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b2e722f40a945f9a0761e9fbd1c0581 : SUCCESS in 2h 47m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7f8990ae55154c1fdc2f70990ec19ba2813c975b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-28 17:58:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"044b9bcfe0adc4a173bd2264bfa58b0d6e22c647","revisions":{"044b9bcfe0adc4a173bd2264bfa58b0d6e22c647":{"kind":"REWORK","_number":1,"created":"2025-02-26 18:28:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/942843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/942843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/942843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/942843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/942843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/942843/1"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-26 11:33:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-26 11:33:12.000000000","tz":0},"subject":"Refactor and add some encrypter unit tests","message":"Refactor and add some encrypter unit tests\n\nChange-Id: I1fb53b4dc9ee38529662e0d4b5ae586990b162aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044b9bcfe0adc4a173bd2264bfa58b0d6e22c647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044b9bcfe0adc4a173bd2264bfa58b0d6e22c647"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
