)]}'
{"id":"openstack%2Fswift~942850","triplet_id":"openstack%2Fswift~master~I69f4eb3d445a32ad98d45b1a8d4be23673c84871","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-28 16:40:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I69f4eb3d445a32ad98d45b1a8d4be23673c84871","subject":"Add config option for whether to run s3_acl-requiring tests","status":"ABANDONED","created":"2025-02-26 19:52:19.000000000","updated":"2025-05-28 16:40:19.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"52fe1bbd7564b16f60f9f88fd1c392e06d5c81fe","_number":942850,"virtual_id_number":942850,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-27 00:57:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-02-26 23:33:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 21:22:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 23:33:39.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-04-09 15:15:50.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-05-14 09:56:08.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"980683d6ff7c0b7cbb0d0d54e6873149905dd5a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 19:52:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70cf73b0b8f38d4669d5ecf02aece5bcdcff8696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:22:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c90cc9b7799c4e8ebcb3650f796b4128\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b52040861c4508a7dd5032fac1f0f7 : SUCCESS in 10m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/717003269a1c4111a7c42fa7a2555ea0 : SUCCESS in 1h 21m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/666f1908a7a2435ab234222a5018de30 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56cf0ec75aa34734a884f927f9ab45c2 : SUCCESS in 25m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2368f7aea8f944e5a70e0d70a027607c : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a3ec7bc1cdb46c4ae8859ec84a33c07 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e31b6454c9c4d2595c2bcd3d7574e84 : SUCCESS in 23m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d170567f7654664aa6fe095723743d6 : SUCCESS in 18m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/70469cbefaa048679098079ea54643b0 : SUCCESS in 31m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/429a4f3b525c415282a4c036b28fb515 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6ef641779f6b43beb7ae2600676ed766 : SUCCESS in 8m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f9886f9558e4aca9d2b9dcc8833194a : SUCCESS in 13m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5982e11547645f3835356b8274c00ff : SUCCESS in 8m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb6c658fc4b045d388255b24cfbf1126 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d120ac58c7354c56b8c5136aaa215c2e : SUCCESS in 51m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/647b1fd876a745bb8ab892e1f80a959b : RETRY_LIMIT in 20m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c0c831c1c20497299bd47cc4c574f91 : RETRY_LIMIT in 22m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b67e04622509446fb54e43250a274d88 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386519525c064b71b56946f690d304cb : SUCCESS in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad70421718894f63bb58f0bd523bd10d : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"83ea038604efb0a7b8609457c686b2b902146c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 22:27:13.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd768368ae884f2bc9052776e6fd17d57032ee87","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-26 23:33:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39c1b0a0a7caf33a7ad86d38cb1fd0df9dc2e653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 00:12:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a769396dcf5d4e8597ba84fa38983ba6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef442748ff384f488668953365d946b0 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":2},{"id":"50b64c98dc2294444ad0d7fc56712e435bbee8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 00:57:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa32ca319401438a990639926d6576fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724c6f5d689a40669ba4af582750d469 : SUCCESS in 8m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba726cd294be40cd83fbefa070d797ab : SUCCESS in 1h 43m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac66f0deccfe40838817e87a0f48f45c : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02352b8b387a4a28801cbd63c1ff8f6e : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76128336fafb4b4884e9ffc83bb40c76 : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5caabb1f9a34ed6899bc025ed0e244c : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eeb45952e4bd4750aeb7fd0a7329e5ea : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/430f7bd448b1423bb335b934da6d800d : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3fac5e12cf334886976b1eb98ad8659d : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a838ef61894d49b1b7e2aac31692a33b : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e9821801bedb47688e55360bb2995657 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7060831648b94c3d8b4ecd3f935f6992 : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63b8dceaecca480692b04ad98d2928df : SUCCESS in 6m 33s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/931d2a900bf046f1a0d41e4e1bac0bb5 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5a4e2063bae4c1e8c8b41b865320f01 : SUCCESS in 53m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8d874ff2c1d4a499d4358238318e25e : SUCCESS in 48m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d74eaab9f19f4c2cbc8b211bef44d6b3 : SUCCESS in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bbb0833ae9748c08cc5de8f0a2112a9 : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b432d4f75ae24711b57ec7e3561ddb6b : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2eb2a61610a542f9b1a14305f768ddd7 : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":2},{"id":"cedbfca1907bb337ab7bab3d60cf2000cc23a05c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-04 21:58:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4845ad1125a31694425557f56a7c90952754c2cb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 09:56:08.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"52fe1bbd7564b16f60f9f88fd1c392e06d5c81fe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-28 16:40:19.000000000","message":"Abandoned\n\nGoing with https://review.opendev.org/c/openstack/swift/+/949716 instead.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22e41fc2ea779f56dce79f4fbd88c3c385c9f5f6","revisions":{"d4d3c644d9ba5119f3b165e40ec2b180c92a259b":{"kind":"REWORK","_number":1,"created":"2025-02-26 19:52:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/942850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/942850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/942850/1"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 19:48:31.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 19:48:51.000000000","tz":-480},"subject":"Add config option for whether to run s3_acl-requiring tests","message":"Add config option for whether to run s3_acl-requiring tests\n\nChange-Id: I69f4eb3d445a32ad98d45b1a8d4be23673c84871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4d3c644d9ba5119f3b165e40ec2b180c92a259b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4d3c644d9ba5119f3b165e40ec2b180c92a259b"}]},"branch":"refs/heads/master"},"22e41fc2ea779f56dce79f4fbd88c3c385c9f5f6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-26 22:27:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/942850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/942850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/942850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/942850/2"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 19:48:31.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 22:26:43.000000000","tz":-480},"subject":"Add config option for whether to run s3_acl-requiring tests","message":"Add config option for whether to run s3_acl-requiring tests\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/942845\nChange-Id: I69f4eb3d445a32ad98d45b1a8d4be23673c84871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e41fc2ea779f56dce79f4fbd88c3c385c9f5f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e41fc2ea779f56dce79f4fbd88c3c385c9f5f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
