)]}'
{"id":"openstack%2Fswift~943055","triplet_id":"openstack%2Fswift~master~I09d0838f8afca90819a1513b36f799c95cccb99f","project":"openstack/swift","branch":"master","topic":"token","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-08-27 23:47:45.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-08-27 23:52:46.000000000","reason":"Change was abandoned"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-08-27 23:52:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I09d0838f8afca90819a1513b36f799c95cccb99f","subject":"Concurrent tests for object controller to use cooperative token","status":"ABANDONED","created":"2025-02-28 19:30:23.000000000","updated":"2025-08-27 23:52:46.000000000","total_comment_count":38,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ab366e39227fac5effa9461bc586fe9d7bcddcd6","_number":943055,"virtual_id_number":943055,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-27 02:44:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2025-08-27 23:47:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 20:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-28 20:44:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 19:03:21.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-04-04 16:47:26.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-05-19 22:02:50.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-06-30 06:00:37.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"eb20d7c39a2420a46ef9f8eba37c0297fda27ec6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-02-28 19:30:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c14ad77b50546e716012992e74710023cb6a233","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 20:30:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/153206db88d5456b8dbe1e6fec9ea5d9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/695d8d4c01f34b1cb501a7a7e81012e0 : FAILURE in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2af6329003558963574445eb4f854c0191c53849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 20:44:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/831561d5f0f343a39a8abd689b88223b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01c9e85a0e64e46998918ab73170911 : SUCCESS in 9m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c5c5e03e3e52489abdeb9423acfbd8de : SUCCESS in 1h 02m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23a1f5aede4a4e40af91b73a46389628 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c1ac6cef53d4f0bbda02e2f7372b62b : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80901cb3ba3f416998370db359d66e18 : SUCCESS in 16m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c865ae054da74aae9ee16dfb511035fc : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a1c09b61c3422f8123f9c4274e6c0f : FAILURE in 19m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/423182b12d744310b91d22b8055c20bd : FAILURE in 16m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/463790d2aba449fe953ed576c9540602 : FAILURE in 24m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2718163d2cf24c2aa974ad50ca47470f : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd97a8a5c8574e5b87f5188d79a1c247 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bbbfa6db1c684016bcc50ae26f86da85 : SUCCESS in 16m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af598776b57b4850b8d472ba65f2f5e0 : SUCCESS in 34m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb9327bee26b4dbf84466b4993c714ad : SUCCESS in 38m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2116051a9f124673a5175157a4c64128 : FAILURE in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1447ec799846cba99b562a9afc34b8 : SUCCESS in 7m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1de77ab582744e979ae72096208f1ca8 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8460c67266c5016464614a1718e154701a5af476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-03 16:46:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f19900381904e33783b264ef8b863d272f99646c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:59:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5aa4a1caa04d0580e8a79e8b0ffa30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5136cdf06d524d7788252ec6b16511b8 : SUCCESS in 9m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab7286870ba440a8add9154a966ef3bf : SUCCESS in 1h 06m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1700ae5bb03b47199d512cf047196c88 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a534e8b53d914897a3f8ad44d066b435 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13a5d237d6b140fca4f787f3a2c0905b : SUCCESS in 20m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2215332e88144833b8a8b0698f29fe16 : FAILURE in 5m 11s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25430331cf79494c959ab6556dc1345e : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0278e327abd44e3b986273c8f8a6b1d3 : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9fdc08cbf3234e9582c788f3c986e57c : SUCCESS in 31m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea43e1d0368e4af68051658b67ebb559 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a10140623a9544e1bf6e7c0c42ac4754 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/349a55686c904127ba4e410df40073e0 : SUCCESS in 16m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15d28e1c416a493e8ac1e3ca629a5106 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c7e7ca66d8643b4939743bc9eccf573 : SUCCESS in 33m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f8b6bc16bc4925ae10ef40fb492c8f : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0befdebc2c4c9ba621f92c8f1f025a : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d20038c831b34bc2aeea4188a7ce6763 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cfa8ad683b57fd0581fa53027783b76bf3a103b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 18:02:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aea4eb497e64ce0b980807abcba1864\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd461c3b78e041b5a227788fa8fd0927 : SUCCESS in 1h 12m 57s","accounts_in_message":[],"_revision_number":2},{"id":"207b6d3d9f78e5d28cb5a953619b38b2f7cd3d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-05 18:36:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1fd08aac3a29ac9971857f021b8a075613f8b1e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-05 19:03:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c14a054b9a2a37a40da3126b6642a577b7c9861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:36:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1ca84dca28a44dfac356aaa4e32876a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca5ada0ccaf49a79b6b7a1892816feb : SUCCESS in 7m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/166445d2918e4ad6836f6ec8f20c694f : SUCCESS in 57m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c73c77b3190437bb2e619e82bce2bf4 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb6b74bcba4f4c9fa1c488b51cc81f85 : SUCCESS in 23m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/926f0bcafd274240bae924dbee9f9731 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c754dd268a284b3d898b1108a90d2377 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9845e18e0edc44ce95572be9e137b557 : FAILURE in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1a85f9c9a747c89628d4d3580693c2 : FAILURE in 17m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/df30420447494220872161c10b35ddf3 : FAILURE in 24m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ad5574e56a79461c94e9c63048be2c18 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/16fd55866de34fdbaf0165e51da9d353 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86a4a9ff00df4ffba46fbd0c58c8ca0a : SUCCESS in 15m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbc844529774418498a4ae57b4b06c68 : SUCCESS in 49m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea9f946b9d88453ab228ba92df10ddb8 : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f28fd4dc1c0d450f8f5aaf335f0db73a : FAILURE in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d87aadac7f47588cf01814bc9559b1 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/641fea9628254f72bf03d42b4e4c96dd : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1e2a5a57dfbd80a9975ba94ab119ecac8ddfb858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:38:52.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b52e9f9b14144128bf0bbcc8bd2235f0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0787402a160f4bba9915b63b627d6b87 : FAILURE in 1h 00m 54s","accounts_in_message":[],"_revision_number":3},{"id":"189db2002d346730ccb7c2336503b2f204fa4a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-10 18:21:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7d640c0b18590572722839cebf83550949d7ed21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-10 18:24:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6dfa782bca7a501a8f3cd45808f2d8a3ce3a7783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 19:05:22.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3a6edddccea40a0aa138f86bbed2ef8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1c3143c3da40229856cbf64adbea54 : SUCCESS in 4m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4435b67fb73e4abaa0f5315528cebc7b : SUCCESS in 27m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d446ec2051545969b9a1635043a9934 : SUCCESS in 12m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb357346b9464da9b50034b51e545f22 : SUCCESS in 11m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddf18ace99d6438b997a18f9a03c1231 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28c2ca421fed41b2945d5afe0acc8097 : SUCCESS in 4m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef5bf33a6c54c4f89f20b05f3397f4d : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db12419d147a4f88a32ec0172ac69b94 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40ecfca2fcff4656870448f624e0e294 : SUCCESS in 10m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/213dadb7f94444d8ad452e1e4fe6cf42 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ed6c498389f74cd495575911a58836f1 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/46f1759581844e28b91ffb404ef06ee4 : SUCCESS in 9m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d30b808b50cc4fcdbf3e749dfd12cb2d : SUCCESS in 37m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93dcb255cb844034bdce456ea0ecee5e : SUCCESS in 39m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1471de8bb4024588bec86db9ef971b12 : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb3dda13cc44647b1766563a2954426 : FAILURE in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39460c11dbab461abd83a275b54db0dd : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":5},{"id":"616362005e9d2adbac420931dffe9670fccf89db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 19:18:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fb534d645d4c11be5c240c6ba70e33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5a7dca4b4a2464d884d7aceca8964b6 : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":5},{"id":"ca83a76d3e25ea98cd3d903b54749183a13a8bd0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-11 01:21:37.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6f759931bde6d9e1b8901e0ddbaa638aa8bb8c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-11 14:49:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"709de4679983b8af6d66973e57a027ec945b4750","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-11 15:10:29.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"56005aabdade827c637264a8c9bf50e399fa39cb","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-11 20:18:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2d2cb43eb15023ebfdf06cdc50954616015300fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-11 22:09:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1911e2d4e4f6efc990be1c917745621da5feb9c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 22:53:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb0b1cd2e2746bd9add30eed6d21008\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1da701a8a70942f1a3165d279a4b1eec : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":7},{"id":"ece3a8f81a60d26e68f934450f48a3a401284886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 23:06:30.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0436aa168819406191eb8253ebe5edb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8892d5c7364441db06aa8d69486f43d : SUCCESS in 7m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/941783f8a6c24a8dabf9c3c6548997d4 : SUCCESS in 43m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfcfcce2bd664a6a972afbb197434258 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43e2baa941f34bae95a55a09be27023b : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd8c8a82c58e41d98f372ae8ffccdcd1 : SUCCESS in 10m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2256e4432dac49a7868bcfaa3d286e33 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7053e2c49e1466ba82ac2cfa5707f3c : SUCCESS in 9m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e96453c807a47d987b2308667505fb2 : SUCCESS in 19m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cb24ac4bf3704023a9d6f5583773551e : SUCCESS in 19m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/96324f7b836b478185974cea6f3428b2 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b77b893b979e4827b94cdee59219efc9 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/10de5434f3104eae834c5332114fbc53 : SUCCESS in 7m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1347b1035ae64d2ab242abcf7ea4cb24 : SUCCESS in 50m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1add0b25d0424efe8435b7bef2f80d13 : SUCCESS in 32m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d9f6d6af4a04354921682e3d4b2374b : SUCCESS in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3292ce00745f4088ae5d0286fefd44d8 : FAILURE in 3m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b55c1056ba76464aaac4032594c6fee9 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":7},{"id":"259ce60ec25c2781fe2ca0d8b3e073d7af39a203","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-17 05:17:53.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a498c1de6967c37d0c22ff530b0a13cfbb454a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-17 22:16:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d0c9e557fc06432836007844367821d5ceb18ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 23:05:20.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947baa99de4943c089dbff1003651d41\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/962dbd0afd094a9887db76cbd37d9b61 : FAILURE in 47m 13s","accounts_in_message":[],"_revision_number":8},{"id":"6be1f8b7034d86ad378d0f29c37046e09924b4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 23:06:35.000000000","message":"Patch Set 8: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1550ffa02a047aa80e23908e63ea0f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954945ff61fb4cf4b688639b14d245e1 : SUCCESS in 8m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5391b02deaf34a0b8a1c9cc7f02740be : SUCCESS in 39m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4973eb238804fbe8ed01d2dedf22af7 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6ca2d9ae93a4c5f84baa85050f17ed2 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fc5764cadd040c6a706e162e76c8807 : SUCCESS in 15m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a20245988224410197b429a9f4cc2049 : FAILURE in 3m 54s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7246e903a8fb4a78a3df89a2befbcb86 : FAILURE in 19m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/badf33c035494cfaac284fcb2263787c : FAILURE in 18m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/13bdf449baa84cbea492133380f5d1d2 : FAILURE in 23m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee8383a1e05f45b2a9d5c0616331b20f : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa7f325d31bf48b382ac80d5af6bc47f : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df3d7792cb4c4f7bb63e7c70988583d1 : SUCCESS in 15m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92c50c5912254b0bb485522aa48f5212 : SUCCESS in 48m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63b9003fcd3e49578b2ec499754c6c4f : SUCCESS in 19m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4852ec3b03d4983bbaed16a044c2a7e : FAILURE in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951ca41eca49407a9c7db10822300c05 : FAILURE in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60b7ba71479848318854438ae07cbedd : SUCCESS in 22m 26s","accounts_in_message":[],"_revision_number":8},{"id":"ada294eea3f912a53a237aba507f3b2722510a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-18 16:42:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"027c536fde0e4da311efb164fc2eb2a07ea29b5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:20:44.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78400fdebf1b477cafd2a53b5973615a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c45ff5b113c4f62b264c8f56119d9d0 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":9},{"id":"9888beccc7ef80f2a0790668d7041056170a775f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:31:02.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fca7b7d2b544d289969e14d60f8154\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2d57cf3b6f4b46afd205122d3dfd48 : SUCCESS in 10m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57f8262bba3d40d0944a62f7857ebc4f : SUCCESS in 47m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6372e546ff8b416b960b9474e4f7f57c : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72741ed5b2914f0e8205a4fe38b57a18 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79179d586fc847da8e36676311f934d6 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fee99c91976045e9abbc054534bbd317 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/baa811c7a701499d895b8a757fc7f7af : SUCCESS in 18m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/76c32349ed76471ea4f55d30e28243c2 : SUCCESS in 22m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b1e953471f2e49dc96a9eb37ec19b500 : SUCCESS in 27m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/649f07baf896479bab988dc44022616f : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/167a79a2183044078dcd45b4771ac519 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e4bf9dd6dc24a1f9670344dde8ee438 : SUCCESS in 15m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4966e1a7080445a4a73d34e3323b6941 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ef2153e7db1450d88cb106ccf83ddb6 : SUCCESS in 17m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6036a104353e412eb2623e27fb8de884 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c3bec073b9c470385f1c01f97f577ca : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f3293e730f54d3c8399dfb52c85f96b : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":9},{"id":"4d9475e16fbe4360c906c331093d4e54770943ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-21 16:26:29.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"80ab358327419b84a3c381b5ff2ae2ab81827ac8","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-21 16:27:27.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ef286776eaf2ef7d6ede2366aa387534a39097c6","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-21 16:40:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c6e03d63542f8c342baca5e09b287f8b5c9710de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 17:15:35.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/984a5a1b842640ecacae03ca75b97f3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d790533394349e2a1661dfb6c3cec70 : FAILURE in 47m 19s","accounts_in_message":[],"_revision_number":10},{"id":"4c1b61b61e7c72d84345ee00182d83c9a8412a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 17:17:52.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6af28c99f25464aab21547fffe2f2ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d909dcc1df461a8b7a748aa9f26952 : SUCCESS in 5m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86f5009e188d4cadb63ca14f5a1ed7b5 : SUCCESS in 33m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6685787a35a743e3b04ef59d7a422af5 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/192a8b74464944638b7de6a645309c41 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24e6f8e2fa5c460589a599ba4ca00c62 : SUCCESS in 10m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbbce1138e014f43944df4d44e230e13 : SUCCESS in 3m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50c6e7786fa249a5a4d4d56bebdf3932 : FAILURE in 7m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a947d9ae19214d33bec27ee2e2813c23 : FAILURE in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/346e247c77c64515b77f4ebafdad76d4 : FAILURE in 18m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/049dc91a31854e8782c55c3e3725649c : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d11e0e69a5094c7cb69547a2e2b277ec : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/74b7f45024de474aa1b6e7b633bed2b9 : SUCCESS in 16m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c065c12ec2e4680960cea60e89797f9 : SUCCESS in 44m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b11a8f3dd6c74f0d879612dd10eb1dcc : SUCCESS in 37m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47af653b3ded4713b33043ee61ef7db8 : FAILURE in 30m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58186f6b68094283a7135bee5cb45907 : FAILURE in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/909b868043a7488caedf654a2aaee578 : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":10},{"id":"d4ad9228c441fb75a77d880c6658ff7bb4898a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-04 16:26:46.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7c9cf2beb90e3240ef3511d0a6eb7aa44972af87","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 16:47:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"839cbea0f0859191511eb25e2acd140fa386c52f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:01:11.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b82602546b0491fb3fd91df5d362a1d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc505ea469ee44d2bc398b2eac42c5fe : FAILURE in 32m 21s","accounts_in_message":[],"_revision_number":11},{"id":"ebf85ac8bdefd4da22c0d9edc9aae4cabb1d6548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:31:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dfc3eca816d4518a5df2f8c5170f051\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74415c88cc034481a30b7e47f7911287 : SUCCESS in 9m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15ef4baa9fab493381e74792a5047d9c : SUCCESS in 55m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1056a6a1a39a458f8692e77c5f31f9be : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5313a35c66184331b816d3d9c7f09fec : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94035d64639f481eba8e9290659aff1b : SUCCESS in 19m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc318bac30c64eb3b1cf7a39889ef2a6 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7ac1731c1346e8a37b0cc0983e1fca : FAILURE in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/320b2f72e58f4b0786879a5e818618e6 : FAILURE in 17m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1e9a154f76cf4c3c96306beb541ec999 : FAILURE in 19m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/829e024653a54e8499871b2bce13bfa6 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a5c5e2935f06451fa8376e8ad0096481 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eee8fa612c76496f9ae5371ce09fcf2e : SUCCESS in 14m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f7af0245ed8434c81b194dd3d25d146 : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75b7943ad56b4ebeb756bc6ecc072d4b : SUCCESS in 47m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/883a5ecb677f43098380c1ab11d321fa : FAILURE in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252e6e8722234132902e802f3b0cad36 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/522a23bceaf6492e80df3a356a5409e4 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":11},{"id":"ca880167d46d38aba118f3b1aa29f70dbf2a5680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-07 02:40:36.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b20f5ee1b60b9f89f9be029c954dd158936b3a16","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-07 02:45:21.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"01c8d922617c8a7ae963ab2dee2612a2c3760c63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 03:28:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847a8aba6aa34d688323a72d92e36720\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e8b9216bc274d2e8228f4366983088d : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":12},{"id":"857ae743671d94599b2625c8a767cbb183e340b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 03:38:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7cf2a03ab14d77b9a8f2033f3a8d97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1665504ddd414098afa4f714af042017 : SUCCESS in 10m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7191f158c9f1457bbf2efa99ddf62868 : SUCCESS in 57m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/344da5341a0e42ed855cce5ffe57eed9 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3448b46159804584962c50d6a993a964 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a5f3c2ced7e4e78a0bc89bf994f0c64 : SUCCESS in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1194c6c189754a11bb3e8c84a0a09cc1 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/15f2a51cfc4549099c3ffc052c3768e9 : SUCCESS in 17m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/300f3baeafe64fdc9c49ee0cc73761e1 : SUCCESS in 19m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/33da6f9c6c44468ba2526350caa0d959 : SUCCESS in 23m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/123201035f5c4d96965c7874701be382 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/94d0e2792f98425ab2c56061dbc55819 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/243176018cae4c1f94798dd8db5e25cf : SUCCESS in 15m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83aec3eb7ead4028ad3c0b61a4d3a681 : SUCCESS in 46m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/043abc134cac466784b4d48cfa81f4e7 : SUCCESS in 49m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/060a69ca1b7a492c9824209d9089c0c8 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87808eec4f8046279dea31e7c0b52236 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c7b138006204d979206259ba0786ddb : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":12},{"id":"f8e82cddf14792e8f60ca3c40049bc596ecd1587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-08 22:40:06.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"10f3e6fbc6ece38d3dcd352eac20abc0bddf9898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 23:20:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/697011c15ef14c51a6484a16b80e0049\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db208fe86bf24493b8dcb27ec93c5213 : SUCCESS in 8m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2317fcb383f4ea89bb1c1a389a2eb05 : SUCCESS in 38m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/743c2b39336c4b42872299287481fef9 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4e56cf576354b9e90ccab2367002b1f : SUCCESS in 10m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74df436c322243d5ae721ddafca3f1a8 : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5a7274210f1417cb0b7afd0a8e8c1ed : FAILURE in 4m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03cb314d4c5f4eb68fee57a98f344054 : SUCCESS in 15m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75fd197a3e714b81b36bf3a94da7ae45 : SUCCESS in 22m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f26800ec87d945eaa31e8630720ff534 : SUCCESS in 22m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a6ea7962925b43739eb8ed87949020f8 : SUCCESS in 10m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f77c6ca3bcd948a1a85a2266cde63be2 : SUCCESS in 9m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/efa7c98023da417887ec3c2fa14b38b7 : SUCCESS in 16m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d09c078462cd409e844b4977691a80dd : SUCCESS in 28m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d8a8fab8eec420eb9a40d1760cadfef : SUCCESS in 23m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3244914e665b448fa8531bc16bced0aa : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f58a63fb8a54a648aa056d984e8ed22 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2215f1e871e348be941c8e93d909e297 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":13},{"id":"c7d9cb2b77b68fd4b5209077fffda4853e011113","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 23:29:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9160ecda57640d69ec800e4870b0a95\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e43958990174a94afc26af55dbb1b1f : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":13},{"id":"d5cda6c25343c0d0e732b73ea69c4d7470595a58","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-09 04:22:01.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"205c3dba4da78d2a40bf6ae359ddcd7b00172abc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-09 04:34:40.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dc3b9f5e393c6ce8debcf6c7f60169481fd7392d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-09 04:46:06.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"81feef8fffbefabda888628f97362c5f80cfafc3","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-09 04:48:40.000000000","message":"Patch Set 14: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"46ee06e5443a22ce389282559b9da0773846b3a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 05:24:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df5c6f75f244487c9fa05c8b2bfdacec\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2fe4b4f7d33404abb9161174a173816 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":14},{"id":"39653974250d7fe0df2ac3e638d9755f68b727f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 05:35:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02afb351244f469888f4e5c7d451a437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56292d421ba64ad9b0947ba8c7fb132d : SUCCESS in 7m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb8600140e664a2286175864d9fef28d : SUCCESS in 41m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2866f494d57f4b5fb7bd4c467f58fce1 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a71a0b8767eb436d903a4c0d71574a8f : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adab20ff3625450684996d65886140b7 : SUCCESS in 17m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6a192d4219784647b24db9821bb71412 : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2626e68f4d64b63acca0293ba42e3b5 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ee87448797dd4fd291d7f413264e1cad : SUCCESS in 17m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/21a86f1ed80b4d97a886bf5b75a7822a : SUCCESS in 17m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/adbc704b7f2a4acea13f5bbe288608c5 : SUCCESS in 8m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b6e7135d72504947bf556576597bbfaf : SUCCESS in 8m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8ea34c6c5960429184c12157ef6a8240 : SUCCESS in 9m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8fece2eca2c04936aa6a7d75c90729e0 : SUCCESS in 44m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1d8e44a25444a6e99af3714a9746839 : SUCCESS in 23m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/911b7e1ac3dd46aa8a2f5525f3f4bc63 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29681f0c9fd0406d913fb9e9abf5b2f9 : SUCCESS in 3m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8c3f6177c164904b1cea8b1811c6dcd : SUCCESS in 22m 12s","accounts_in_message":[],"_revision_number":14},{"id":"1c6b3de04ced7d0c1cd134a5bfadafaa88642f69","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-09 14:37:05.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"fc2d63855fd7b6543841ce26b9a1fdab5ef796a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-11 00:04:48.000000000","message":"Patch Set 15: Patch Set 14 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":15},{"id":"f00069fba52665b1539d933c8d0902cbfeee1930","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:44:32.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c6ad28627148278f61a3e020a67726\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b267f537fc80451aa7db415a1294dd5a : FAILURE in 37m 46s","accounts_in_message":[],"_revision_number":15},{"id":"c9e44e1789598f36971f46acad94a915655994c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 01:01:12.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/018fb94a44de40118a2886f52b601616\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c853cb3ac294a018b6280d51a7d8898 : SUCCESS in 9m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f1473d92d3f452a8a78dbde2b95ac99 : SUCCESS in 55m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5a63e76211a43a5892be102ebe32ee1 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f54482419be447a96015aa74a9098b7 : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46e209f1ab84484f8d72cc3e18bb0ff1 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f30e125c09754289bc0c4065b000580d : SUCCESS in 3m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f8e1d9757d6456a84c5144562fedc42 : FAILURE in 7m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/389b20a1d7fe44bf9c3ec4efbfe423f6 : FAILURE in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f438263be2e04341920e501e7b29210b : FAILURE in 21m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/40ad546dea8f45299f3c5095507d27ff : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5ba5d47003f74bc7b2b2d2a22c017cd7 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f31be9cd5a3e4b1595bfa746fa0a4d41 : SUCCESS in 9m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/886a28f292ab4a0e90636b2de57906fa : SUCCESS in 49m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9017ffcfe2c8468cbb93b222a97f3762 : SUCCESS in 50m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/681428ec91524b35949c3495b130b748 : FAILURE in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998e8ff693be46ae94e80e9fd2e6ff66 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2788b09159d7491787e829e3af070703 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":15},{"id":"9648216497f11d954257913dd10ee3a8fd5a8b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-11 16:18:45.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":16},{"id":"2a6b2828ca44c34b40005027c441b070f5f16120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 16:53:13.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d3fd69cd251436ca8b5e203eb8cc86c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ee331d205ac414baa7abf3b0215fbe1 : FAILURE in 32m 34s","accounts_in_message":[],"_revision_number":16},{"id":"aecb5410b99eccb71ad9474ed433817e6db46618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 17:06:27.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75bcd0de6c3b4ac0b140944432924c24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d679e3fd33343e4bb5c231330ddf1a1 : SUCCESS in 8m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37e2d07fa281489e93c547434a2449ae : SUCCESS in 44m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c86061d7c3b482dafa6f55dd29d452c : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6fc043d59c043e0b84126ba0d0629c5 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2f6f9074553472c93b57aec3da1566e : SUCCESS in 10m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ced2b1bdd954b19a1ef76fe60a82e1a : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a373a5dc76154e7c9bc868b3a103260a : FAILURE in 7m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c6f45737c5f4cca840b872f471bef1d : FAILURE in 16m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c402c16bd0b1416a9f843a0946d2d465 : FAILURE in 26m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76691eddee6a45a8adcb89843985f503 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e4d9a912010b48988911a3d5e7486acc : SUCCESS in 15m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1519574031434dcc8befa31e41edcc83 : SUCCESS in 15m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3ad3ee26ea949bcb2fd8fca6e84a040 : SUCCESS in 38m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb9f5471a5db49a2afb61846f9423692 : SUCCESS in 42m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f601899ceca24f6a8f8d93514cb3021e : FAILURE in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d537b8ac558499788de0eb6546dc1ce : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b878b8a47284baf8bd7c92ab0525916 : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":16},{"id":"756cb94a33e528bbf954c1ec4c36a9f972251e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-14 12:16:46.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"6f063c3a0d269488933e909ba3eb0c2f1656fc93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-14 12:31:39.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"2be5453c43d6c8d90a5e3737c821364c02d3101b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 13:06:12.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5023f69e84eb4f849c420381d070704d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eab6430c88df4f928da08009506a2839 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":18},{"id":"9f0335e86ed61426b285a87795d87054d0ead5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 13:37:08.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06debc3cba634a78b7c5be2f3e8f9b4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85100419ee404bb9a2e044f2be7b6b03 : SUCCESS in 4m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f78a5f43bccb4a96bac4d729ccfe4f19 : SUCCESS in 59m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83784ea41b464c6799efdc71604ef225 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77e3ac31959a4e9da105ccedf8d466ee : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eab8270910542fca5fa71f0c4d77914 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/039646b69f2847b0a93edc8f6d947872 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/037f93baa8544a7bbac8e35ce55e4506 : SUCCESS in 8m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf67049001d490cb21a1e09be904edd : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1a428f68ed44837af5e76accbb797f0 : SUCCESS in 28m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de5780242935439ab94985a917177caf : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/746fd944e4904d749aee300962a31925 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a7bf9a50b0a4605a0d3bbd0fddc50fd : SUCCESS in 17m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb6f48d16d524c21b1351c1639faa8cb : SUCCESS in 37m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7b2fbb8c3db4c51a39642ce6bbdb721 : SUCCESS in 44m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a6faab244484ca3a054dccb00b87f46 : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e068f86aedcc4eafaf10b611b561f083 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9dde10021f3b4735ba41c7fe9eed8f4f : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":18},{"id":"c83de768969856bbb8fa927ebfcf11581912f2ec","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-04-14 15:31:45.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"ea1cd409f12632278cf89335dac8e1865bd0be30","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-14 16:06:24.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"503db87995b4183df31cd57dde534dd8793e1b53","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-19 22:02:50.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"34e3d0caebfb0d0ef0f331ca97dc08445796fc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-06-30 05:59:45.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":19},{"id":"92ef56740ebbbe8c0b3dcea990cce86db1eeda2e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-06-30 06:00:37.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2240e21c8b86bed0f6963d4741a7e34b02e2b11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 06:45:17.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1253448d546f48a6b6c3ffecfc36dc99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477c088eb98b48dda5e87f29eace75c4 : SUCCESS in 7m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae8ed8bf1c3340728bb261775d54efc8 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25c635361c9342288e86b29c8cf23440 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cc7def7114342038a2be381149d4c2c : SUCCESS in 14m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/61ecc5a027db4918b7a3403de47a75cc : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc0201ef0d3d47b59e570199d391d0c8 : FAILURE in 12m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2e7771102764c809a32b220ef208f81 : FAILURE in 15m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d570dc8102464e0fafc8cf413ea0052a : FAILURE in 10m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8af0985438d848eca4f58eea2a3237ca : FAILURE in 39m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bf7678d915c049e1a225f4200c6349af : SUCCESS in 12m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ccaa8ddb730b4c4e9a86ec928b75125a : SUCCESS in 7m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5038c84fac94b559422b1f508a3f4d3 : SUCCESS in 15m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5b36ce66ad44533a33f7abf20be1e04 : SUCCESS in 32m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9acf052cca6b41ef8171d29a9368a0c0 : SUCCESS in 19m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/140002cdcf06434893eb2908dfb6eff6 : FAILURE in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c16f51437ce4cf28c63b6b6c7e970ff : SUCCESS in 2m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56db662a01824f79846a171314c08d0d : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":19},{"id":"677478fa0f11b3e504c68ec155452a7c99a1049a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 07:13:36.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7193853f612b4ebb9c15c2dc00bd79b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b91a0a09eb464409ab1c25576b5d146c : FAILURE in 1h 11m 12s","accounts_in_message":[],"_revision_number":19},{"id":"c3b4eec8e81b72750f1dee241a3b1de64ad54197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-13 16:11:32.000000000","message":"Patch Set 20: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"89014a2fde5fd839aa18e9db00bd0bb2628f9533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-13 16:11:38.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"ca728b84ca1a12c55626dad8a9b5807a13f2ffe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 17:08:14.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/793e298bfc374c85a0095ff978a0cbfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bad380abd54167854da773d07d151e : SUCCESS in 12m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32f71f1a3afa4fdbb6073b3ca0547c67 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89d4a4363fde4fad9d05ecffe211dd22 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a2d8c5817c14857be38a50a0e8082de : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aec164958cea48e5867d85262865018b : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f6391b9a4764113ba2fecafa6f65eb0 : FAILURE in 19m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b08fecac40ce4f4a974e81d07b1c8beb : FAILURE in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b2878c417ab42e9906949948189e540 : FAILURE in 24m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb2359b298434b77be7eadf1d2508de0 : FAILURE in 24m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/81abeb77274c428cb3219b03c857a55d : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9a579b0df799472687ba72130e5c8aac : SUCCESS in 16m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/31d5969aa27e43e8a834d5b4a860bec6 : SUCCESS in 16m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d146cba44044f55b25f203f687b8104 : SUCCESS in 50m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e60c8e307d544cb2a3a1cb4b27f2341d : SUCCESS in 43m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abd2ea219fb64ab193364bb371620ba4 : FAILURE in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d841293342d492491cb6d178648c068 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ce63f11a0bd4dd9864267d612fe00c5 : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":21},{"id":"68fa9e518792171ce8d38f761f7806e941238078","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-13 22:39:22.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2b00157be58e7617d5a5f1d2e9f08f548966061b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 23:44:22.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/707f749ed3fc4b9e97efe8da54113677\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9088ae657f2645cc9e7d6f50db517805 : SUCCESS in 9m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ce3c75147a2406eb13deb593ef3d8d2 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35f9423444f24645865041fdc4253229 : SUCCESS in 11m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6177b148f8ee4ca893257fd487d2fdd1 : SUCCESS in 10m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9768c91176774ed6b32acd55e9fec8f0 : SUCCESS in 3m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62a048fa027c4300a8b725541501be10 : FAILURE in 14m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/136f43114fb847998b2b60896232f7c6 : FAILURE in 10m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea0d936160dd4172841f1be11fc57dc7 : FAILURE in 12m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a5ad8753b4e74feaa33db083d5c5d71a : FAILURE in 32m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/827fe985194d4806b664ccfd5e304d51 : SUCCESS in 7m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bc496fe4d6414346a61bff830aff63f3 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4d93bac963b44d60beb90fc177086fc5 : SUCCESS in 15m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/338b14807d2e4c2f8e79159f3741adc3 : SUCCESS in 54m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83dd94ca69834216aa7ca195b455c862 : SUCCESS in 21m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55f49dc65eb464793e3c933379b41e5 : FAILURE in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b0fcab9c5a43d59c11d72e80224235 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e51339f9ea4450fb7561532ca1be790 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":21},{"id":"93f5bab34e604cffada6dfddd9bf21ee2cc0e5bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-19 00:18:17.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"fb46bd8967b84eaa01ab666ffdcc46f2dd433f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 01:16:08.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6b2a6d855ed4a1b982ec20b8799cec5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771fb7259420439eb769255b3e954a59 : SUCCESS in 10m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34c8a137523c47aaae7b62b41b344385 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa8efbc544aa4fdea1b92f07cbe58c5a : SUCCESS in 10m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7072324a40641f9be39c083e21d179c : SUCCESS in 10m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f47588d4cef45afbaab9afa655914ec : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e553342fca404bf58d01dac97c035f19 : RETRY_LIMIT in 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bff5a2a23e14ca7a8c48a3f46560426 : FAILURE in 15m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9b22ea5ff5a94a3b858c6378c85e95f7 : FAILURE in 23m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2db4b38c69154335af1cc0f0d8be6583 : FAILURE in 14m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/99d10465d5054b319710c64cefe26be6 : SUCCESS in 8m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c3bae0d888114317b228566b9d77bf5b : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dad65a280cfe4e2d8f5f5e3b2e59f178 : SUCCESS in 15m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f96670e485e24cf9ba126b15132c0cff : SUCCESS in 55m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad18f1d5908e494c892d4187c49dc94f : SUCCESS in 25m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfc322a2bdec42ab8243978c5dac89c8 : RETRY_LIMIT in 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df2d478e8a84a7698e510d14a07c410 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d048df6ce4f498b9f30edb61f839120 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":22},{"id":"19df4e795711688c5abb38aec5fd1cb6eeb8935a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 01:24:24.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ed89fd62d045b0beab77aa46d1c670\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d68f331310cc4ae8b92b686c8ee965b6 : FAILURE in 1h 03m 45s","accounts_in_message":[],"_revision_number":22},{"id":"18605a1503fdc870080fef230d074492a7ed82ec","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-22 22:11:20.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"f5334f632f4136e1fa236072e81a7e4a6f3fdda9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-22 22:28:52.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"aca96b09e27d401f6e9641d4ec915b5a70250e4e","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-26 23:31:20.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"f499e94cf17a429186be7301b713afb4d8000728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-26 23:37:07.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"e384a1383523a2a812d2a10186031f62839ab6cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 00:30:05.000000000","message":"Patch Set 23: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79a841d28440460eb058a6fd6c97e70b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6535d08a67854a9fb193232095784724 : SUCCESS in 10m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/542cd71796174f14a92d275a8fc360b9 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5057513fc33b47f1a2f7854427b3cc41 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56ce3321d6174df190c05786d8da36e2 : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecceb434d20b41a49bac36324dbd8d0a : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bec5e2aaf50e45a7b0c36dc8f6792024 : SUCCESS in 13m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04d036b88bf244e4b644e5e3c5827d0b : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/687a848a774441b49c0b85b4c61f5726 : SUCCESS in 10m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d5108ed9d8ed44ccaa69d0ad40ac73f2 : SUCCESS in 23m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3b582f1a1ee6482c97dd75c4745d986f : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/958ac249f5f34fefabdeb62e4bcdb420 : SUCCESS in 8m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/55a96f9272924c1081c8f9a3ed009dc2 : SUCCESS in 8m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/989834f64d0d4bc5b8f1465962ac9232 : SUCCESS in 26m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/294707cf51ed4663af92d23c170b920c : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d3702983a074877a7f7644f9ab086a4 : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb178f7385b4ef6963c6c822455d01d : FAILURE in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1a2fca70d984164a707c505a7a143dd : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":23},{"id":"21719b05efe96b1f6485e5d7b99ba1e5eb76901c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 00:36:27.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d93dfe594ca442ef8bec79298652f059\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e384701548e407895efaf4847dd8cee : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":23},{"id":"a72ded7b23f5cc02627fd30fa28d5b6e4f5f513c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 01:47:49.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"f6381b4dffc8c2220761aa8e989a8aeb66a189dd","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-08-27 02:09:12.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"c7d7e70fbd3d723bc1001cf714c44938885f4154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 02:44:54.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8debf4cc5724d3591d64622e1e00973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237cb4b9cec64169a89b52ee9636a102 : SUCCESS in 5m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/221782e4ba0d48478f84607f10162468 : SUCCESS in 11m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6dd0ccc9c9d491eb41f730d0e3d8e04 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07af69b46816432b903dcd6d1b62144a : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66c1251eb50d4270b88d45d97f225a57 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71d034987ef044a7989cfee5f3a0c754 : SUCCESS in 28m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a00f26c2dcd243d2ae721c21c75f2cdb : SUCCESS in 8m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b117bf0484724c7e9893fee637d2119e : SUCCESS in 19m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bd23cc2377174fcaba20a214c2dffeba : SUCCESS in 35m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/076d5862947241c3a1cc6ed253668c48 : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bde87d8caad3401cac18edd12d1b2edd : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53a8605b1d204b79b7efede636afd7f8 : SUCCESS in 15m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c832d9effd174b52a24086f8043e663a : SUCCESS in 48m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77b75be76ec34028842db40301491e52 : SUCCESS in 52m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1deebd1249a24f5dab33cf61d2f1bcc5 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4139f48c7e5d452fb216ced02d79b6b1 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b1223ce7d81403da60985f3a51bdeb2 : SUCCESS in 23m 44s","accounts_in_message":[],"_revision_number":24},{"id":"cb9ca8e49eca41aa1b1412c602af68214403fe82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 03:27:07.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6db518d2cc8e4bc983856178a624e5f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d3a44c561d14dfd80d49dbfda7acea5 : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":24},{"id":"7c3d96ba825a40e0677e69d1ab8bd7cfc2c53ee0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-27 23:47:45.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"ab366e39227fac5effa9461bc586fe9d7bcddcd6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-27 23:52:46.000000000","message":"Abandoned\n\npatch squashed into https://review.opendev.org/c/openstack/swift/+/908969/52?usp\u003drelated-change","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"ec996e312722d2d2b18dc82b736ea688007901ca","revisions":{"96233abb260b458848f7b8a4ed5b1ad2694087b3":{"kind":"REWORK","_number":1,"created":"2025-02-28 19:30:23.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/1"}}},"commit":{"parents":[{"commit":"3b2b8859917b8aad03423f082f2f6a7c7b48ea9d","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b2b8859917b8aad03423f082f2f6a7c7b48ea9d"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96233abb260b458848f7b8a4ed5b1ad2694087b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96233abb260b458848f7b8a4ed5b1ad2694087b3"}]},"branch":"refs/heads/master"},"ec6f9f56819cf26e8aae8ef7c8675cde41bbd6f3":{"kind":"REWORK","_number":2,"created":"2025-03-03 16:46:14.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/2"}}},"commit":{"parents":[{"commit":"3b2b8859917b8aad03423f082f2f6a7c7b48ea9d","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b2b8859917b8aad03423f082f2f6a7c7b48ea9d"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-03 16:45:46.000000000","tz":-300},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec6f9f56819cf26e8aae8ef7c8675cde41bbd6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec6f9f56819cf26e8aae8ef7c8675cde41bbd6f3"}]},"branch":"refs/heads/master"},"e737de5e64204f722e200a38cdc4ff23da6133ea":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-05 18:36:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/55/943055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/3"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-03-05 18:36:03.000000000","tz":-480},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e737de5e64204f722e200a38cdc4ff23da6133ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e737de5e64204f722e200a38cdc4ff23da6133ea"}]},"branch":"refs/heads/master"},"e3a546dd42bb38e9b898bf420e045da7c0db73cf":{"kind":"REWORK","_number":4,"created":"2025-03-10 18:21:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/4"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-10 18:21:14.000000000","tz":-240},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3a546dd42bb38e9b898bf420e045da7c0db73cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3a546dd42bb38e9b898bf420e045da7c0db73cf"}]},"branch":"refs/heads/master"},"000001ed034cbdae002db286f398d43abe6ea28b":{"kind":"REWORK","_number":5,"created":"2025-03-10 18:24:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/5"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-10 18:24:19.000000000","tz":-240},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/000001ed034cbdae002db286f398d43abe6ea28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/000001ed034cbdae002db286f398d43abe6ea28b"}]},"branch":"refs/heads/master"},"3a5e43c2c7c32a8c8a350c3fc156b9c63f6be247":{"kind":"REWORK","_number":6,"created":"2025-03-11 14:49:51.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/6"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-11 14:49:24.000000000","tz":-240},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5e43c2c7c32a8c8a350c3fc156b9c63f6be247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5e43c2c7c32a8c8a350c3fc156b9c63f6be247"}]},"branch":"refs/heads/master"},"88bbf8f3faa0f9d7e535292076c815130adaef86":{"kind":"REWORK","_number":7,"created":"2025-03-11 22:09:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/7"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-11 22:09:02.000000000","tz":-240},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88bbf8f3faa0f9d7e535292076c815130adaef86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88bbf8f3faa0f9d7e535292076c815130adaef86"}]},"branch":"refs/heads/master"},"8b75d1f3d7d34c19cca603a9862f940a1177c5db":{"kind":"REWORK","_number":8,"created":"2025-03-17 22:16:16.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/8"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-17 22:15:56.000000000","tz":-240},"subject":"test for cooperative token memcache","message":"test for cooperative token memcache\n\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b75d1f3d7d34c19cca603a9862f940a1177c5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b75d1f3d7d34c19cca603a9862f940a1177c5db"}]},"branch":"refs/heads/master"},"0171dc629d982ca4d709bd1fcbc1a1ad67054279":{"kind":"REWORK","_number":9,"created":"2025-03-18 16:42:02.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/55/943055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/9"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-03-18 16:41:51.000000000","tz":-420},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0171dc629d982ca4d709bd1fcbc1a1ad67054279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0171dc629d982ca4d709bd1fcbc1a1ad67054279"}]},"branch":"refs/heads/master"},"8079106d66e4e4fe29d208b18efbb465455b48f7":{"kind":"REWORK","_number":10,"created":"2025-03-21 16:26:29.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/10"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-03-21 16:26:02.000000000","tz":-240},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8079106d66e4e4fe29d208b18efbb465455b48f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8079106d66e4e4fe29d208b18efbb465455b48f7"}]},"branch":"refs/heads/master"},"ec8c617843aebc0eb097456a69104929245d6dc9":{"kind":"REWORK","_number":11,"created":"2025-04-04 16:26:46.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/55/943055/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/11"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-04 15:44:12.000000000","tz":-420},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8c617843aebc0eb097456a69104929245d6dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8c617843aebc0eb097456a69104929245d6dc9"}]},"branch":"refs/heads/master"},"da12436be8863f4f0f15fb4209d14fbc85783d52":{"kind":"REWORK","_number":12,"created":"2025-04-07 02:40:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/55/943055/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/12"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-07 02:40:05.000000000","tz":-420},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da12436be8863f4f0f15fb4209d14fbc85783d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da12436be8863f4f0f15fb4209d14fbc85783d52"}]},"branch":"refs/heads/master"},"be331209e7a798e6d319ba02bb36e546d7191e50":{"kind":"REWORK","_number":13,"created":"2025-04-08 22:40:06.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/13"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-08 22:39:40.000000000","tz":-240},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be331209e7a798e6d319ba02bb36e546d7191e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be331209e7a798e6d319ba02bb36e546d7191e50"}]},"branch":"refs/heads/master"},"2e38c5f43bbaf0a5da3d99a8385e84dbd0c922c0":{"kind":"REWORK","_number":14,"created":"2025-04-09 04:46:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/55/943055/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/14"}}},"commit":{"parents":[{"commit":"7a06c3ea3f1eac152244b3c029baa773bb00f639","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a06c3ea3f1eac152244b3c029baa773bb00f639"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-09 04:45:49.000000000","tz":-420},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e38c5f43bbaf0a5da3d99a8385e84dbd0c922c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e38c5f43bbaf0a5da3d99a8385e84dbd0c922c0"}]},"branch":"refs/heads/master"},"38ec1a2e2ad2d86713331567d4f3db4092e3e4fa":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-04-11 00:04:48.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/943055/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/15"}}},"commit":{"parents":[{"commit":"e10de09176a7ac5e75e2e2100ac5353214aa184d","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e10de09176a7ac5e75e2e2100ac5353214aa184d"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-11 00:04:48.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38ec1a2e2ad2d86713331567d4f3db4092e3e4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38ec1a2e2ad2d86713331567d4f3db4092e3e4fa"}]},"branch":"refs/heads/master","description":"Rebase"},"7dafb8b097b7f0d2ee332cb9b3a376d619b00fe5":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-04-11 16:18:45.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/943055/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/16"}}},"commit":{"parents":[{"commit":"3136ea74d3bc9a03b0553fec387cc1411e4e80a9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3136ea74d3bc9a03b0553fec387cc1411e4e80a9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-11 16:18:45.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dafb8b097b7f0d2ee332cb9b3a376d619b00fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7dafb8b097b7f0d2ee332cb9b3a376d619b00fe5"}]},"branch":"refs/heads/master","description":"Rebase"},"f2067e956f078a1f32c378d4f028aaa892a55d42":{"kind":"REWORK","_number":17,"created":"2025-04-14 12:16:46.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/17"}}},"commit":{"parents":[{"commit":"3136ea74d3bc9a03b0553fec387cc1411e4e80a9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3136ea74d3bc9a03b0553fec387cc1411e4e80a9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-14 12:16:25.000000000","tz":-240},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2067e956f078a1f32c378d4f028aaa892a55d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2067e956f078a1f32c378d4f028aaa892a55d42"}]},"branch":"refs/heads/master"},"b1ba76f1de3ab98f40d5f8923f230120f5d221cb":{"kind":"REWORK","_number":18,"created":"2025-04-14 12:31:39.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/18"}}},"commit":{"parents":[{"commit":"3136ea74d3bc9a03b0553fec387cc1411e4e80a9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3136ea74d3bc9a03b0553fec387cc1411e4e80a9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-14 12:31:22.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1ba76f1de3ab98f40d5f8923f230120f5d221cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1ba76f1de3ab98f40d5f8923f230120f5d221cb"}]},"branch":"refs/heads/master"},"80c1af9b3fc8e47e9004dd7a82c362e6d92f8432":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-06-30 05:59:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/55/943055/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/19"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-06-30 05:59:45.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80c1af9b3fc8e47e9004dd7a82c362e6d92f8432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80c1af9b3fc8e47e9004dd7a82c362e6d92f8432"}]},"branch":"refs/heads/master","description":"Rebase"},"9f305eb64a300de88e08bb2e134cc5e5a17d66b8":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2025-08-13 16:11:32.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/20"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-13 16:11:32.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f305eb64a300de88e08bb2e134cc5e5a17d66b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f305eb64a300de88e08bb2e134cc5e5a17d66b8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"92e59b8a88c6561d43e285fdeda32838320c8849":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-08-13 16:11:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/21"}}},"commit":{"parents":[{"commit":"52aacddfd68ef4d149bf40010f3b66587aebada9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52aacddfd68ef4d149bf40010f3b66587aebada9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-13 16:11:38.000000000","tz":0},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92e59b8a88c6561d43e285fdeda32838320c8849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92e59b8a88c6561d43e285fdeda32838320c8849"}]},"branch":"refs/heads/master","description":"Rebase"},"526b65a41fcc36faaa8b3ec1d1583fc0058121f3":{"kind":"REWORK","_number":22,"created":"2025-08-19 00:18:17.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/22"}}},"commit":{"parents":[{"commit":"52aacddfd68ef4d149bf40010f3b66587aebada9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52aacddfd68ef4d149bf40010f3b66587aebada9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-19 00:17:52.000000000","tz":-300},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/526b65a41fcc36faaa8b3ec1d1583fc0058121f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/526b65a41fcc36faaa8b3ec1d1583fc0058121f3"}]},"branch":"refs/heads/master"},"69fcfda70fd9f12b6da8b73a49b37c7478859a3a":{"kind":"REWORK","_number":23,"created":"2025-08-26 23:37:07.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/23"}}},"commit":{"parents":[{"commit":"52aacddfd68ef4d149bf40010f3b66587aebada9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52aacddfd68ef4d149bf40010f3b66587aebada9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-26 23:30:38.000000000","tz":-300},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69fcfda70fd9f12b6da8b73a49b37c7478859a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69fcfda70fd9f12b6da8b73a49b37c7478859a3a"}]},"branch":"refs/heads/master"},"ec996e312722d2d2b18dc82b736ea688007901ca":{"kind":"REWORK","_number":24,"created":"2025-08-27 01:47:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/943055/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943055/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943055/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943055/24"}}},"commit":{"parents":[{"commit":"52aacddfd68ef4d149bf40010f3b66587aebada9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52aacddfd68ef4d149bf40010f3b66587aebada9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-02-28 19:12:38.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-08-27 01:43:09.000000000","tz":-300},"subject":"Concurrent tests for object controller to use cooperative token","message":"Concurrent tests for object controller to use cooperative token\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/908969\nChange-Id: I09d0838f8afca90819a1513b36f799c95cccb99f\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec996e312722d2d2b18dc82b736ea688007901ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec996e312722d2d2b18dc82b736ea688007901ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
