)]}'
{"id":"openstack%2Fswift~943355","triplet_id":"openstack%2Fswift~master~Ice38af6427fb6f30e07159a7771f07415f4d223b","project":"openstack/swift","branch":"master","topic":"https-dsvm","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-05 01:55:56.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ice38af6427fb6f30e07159a7771f07415f4d223b","subject":"tests: Accomodate some apache2 quirks [2/7]","status":"NEW","created":"2025-03-05 00:40:23.000000000","updated":"2026-02-10 23:15:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5be2ac6c8e36bb9052c423396dd7778518291742","_number":943355,"virtual_id_number":943355,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-10 23:15:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 01:55:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0332c7d4d32854a7ef7f5976f2944802a7a61db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 00:40:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0182a7ca8198fff4cc0aca44feaaf1ca5df1544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 01:55:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd0ea9664acb44cca5794195ae7f4f66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c32146f92042eb84b8fd072ac09711 : SUCCESS in 8m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/564b6de8d6774218ab73257885facfa0 : SUCCESS in 57m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99fa9ded8d3c449f9a069eba44aaa8b8 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afabb212b1b3469a80831384c4e63c44 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3232dfce640146f8b6f2ea2477097086 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/53054ffeb03b40169759be362e049a5b : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a987397b85b043e8afd4d219378a1abd : SUCCESS in 16m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d00feaf3bdc94531b7e1206734dda40b : SUCCESS in 17m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb4f2602ec694fe58061d714688a50e9 : SUCCESS in 32m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/228adf92a6a64f2886fb22682ad009de : SUCCESS in 47m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dad6ea1c4343a6bc6b43bc8b2c17ba : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c2b8efd4432418eae3ec9be0553094c : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"989663c2007ec3095640941ee98fd7679150c284","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 03:04:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f171bb110c4a4f86c0967a0c54b6db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/996675a3c8df498695705ab82e9af6c6 : SUCCESS in 2h 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4b1fd1e1918e346256035febaa6d558b76bc782f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 21:50:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b73e297335c3faec9dd0db277216f5e02cf62e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:00:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb9412f3516415b8b67c119cb5a0d5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12391eb5adb84aee8d5ff90587e19735 : SUCCESS in 9m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb84a8ac5c2f42ed85ca29486902efb1 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3c1657bf56a4345970a2381d9ec04a6 : SUCCESS in 11m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38b92af1039940b4bee67b860cb2d8fd : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a1864dd19564038969b076124d46651 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a6d97ae443b4b9e9ad3bdfaabae3a1f : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b48e226e8d684ba382a8c0e538af5f67 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bec12a435b64409cbeacca1dd81f78ad : SUCCESS in 13m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/971d196703b6480aba7cf8840298c9be : SUCCESS in 32m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c83ea76ad3c44c3a3b3ddcd72d4415d : SUCCESS in 35m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081f9f1dfb17435eb98ba4064f74f2ac : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76111540a76e442c852e078ce8e23d7b : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ca535582464f49781a2a9374b3ea393dabdc6298","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:07:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41eda87df5245b195097bac5756b1a6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27a99fa79e11484db25054081362d0b9 : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f39f43a9ef5e058675958583ee7bc8097b5958b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 22:28:34.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a51d3f2235d65b21e1042b8df094f3b587f4950b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:37:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2420fda06d46a287eaa9909c6924a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9f5e9429744ede9e730ff3067fc7fb : SUCCESS in 8m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34007f7e738649c19e2bf37a061826ff : SUCCESS in 1h 01m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d1a44c190634811b63ee23bf7a89b51 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d602d08e96e48a8ad6006232f1fcdf5 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1e9ac9fa3554eb391989bd49970e86e : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e0cc7f9ea49b432b8fd63ce32eee5f2a : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d9d2294a6a6e4d97a1d627fcfd27ed09 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3cd8baa6f2484018ab65e649126ea2f0 : SUCCESS in 15m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08d8730158c746cc8594f68f6b25fe96 : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/233b6db4f6a744ac852cdbdefb449d7e : SUCCESS in 52m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48188020700f40c3a4e543cb425896ea : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46026548801d4d7fb9f394c72b496157 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":3},{"id":"55a68e40715c45a4382093af52e4c437038dc4fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:32:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6a4218f8e641e88fc09300c02bafc0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ae8e25fdd5644f9858675f15a6909fe : SUCCESS in 2h 01m 15s","accounts_in_message":[],"_revision_number":3},{"id":"97c5364558c0a446e4e3515cbda8d2ecc00affd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-10 22:18:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e5c93092069300477e9f23474e20d7744a66b2ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 22:56:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c01c9b9cafd468194f6cac4d5b8c675\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2312d48b96447958ee5f3e2fc65fe11 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":4},{"id":"5be2ac6c8e36bb9052c423396dd7778518291742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:15:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061f3886645b4749b58d16a549916fef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4040d54f04024bb7aeecf99b21f55d2a : SUCCESS in 8m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2ef3ff15cb241d3a2dc1478ea2eb312 : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9e1a859db6e46b5bf66191480bb9501 : SUCCESS in 23m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fd02568fd5e42578ec9079cfce28e41 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8bd633d6d2334efdaa4a4be13abc3777 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7d61b88b936e468298144406f77a96f6 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/361d251fa47141a3b90c28dcc21fce20 : SUCCESS in 14m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cc5ac357f2f45d9b3da14de8bc408a5 : SUCCESS in 49m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25403f558c244ac29a92568142596ec9 : SUCCESS in 55m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894a848e425c48ac96f4a93a71ac9499 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09826010567840ee9fecf5a6af685ebf : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f19004eb2a2928a34ff31def79b9dbe282bb86d1","revisions":{"2b58014b8a39a73d7a37228e81f080d0fec60b00":{"kind":"REWORK","_number":1,"created":"2025-03-05 00:40:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/943355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943355/1"}}},"commit":{"parents":[{"commit":"e8f5117c46898b29045e48788bb2097e8c258073","subject":"tests: Accomodate some apache2 quirks [1/4]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8f5117c46898b29045e48788bb2097e8c258073"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:43:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 00:38:17.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [2/4]","message":"tests: Accomodate some apache2 quirks [2/4]\n\nApache2 is willing to drop the bytes on the floor on behalf of clients\nif we don\u0027t 206.\n\nChange-Id: Ice38af6427fb6f30e07159a7771f07415f4d223b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b58014b8a39a73d7a37228e81f080d0fec60b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b58014b8a39a73d7a37228e81f080d0fec60b00"}]},"branch":"refs/heads/master"},"d66cba7a7253d18497981808fbe96100a760a5e8":{"kind":"REWORK","_number":2,"created":"2025-03-05 21:50:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/943355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943355/2"}}},"commit":{"parents":[{"commit":"2b3602b9be80322702ef6af2f0d792bee70c5999","subject":"tests: Accomodate some apache2 quirks [1/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b3602b9be80322702ef6af2f0d792bee70c5999"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:43:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 21:48:58.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [2/7]","message":"tests: Accomodate some apache2 quirks [2/7]\n\nApache2 is willing to drop the bytes on the floor on behalf of clients\nif we don\u0027t 206.\n\nChange-Id: Ice38af6427fb6f30e07159a7771f07415f4d223b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d66cba7a7253d18497981808fbe96100a760a5e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d66cba7a7253d18497981808fbe96100a760a5e8"}]},"branch":"refs/heads/master"},"543ab8ca1032d8f84d8c009c13447824282d408b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-06 22:28:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/943355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943355/3"}}},"commit":{"parents":[{"commit":"e5e86891bd8c5509bbf7e0f4828d1610f8007557","subject":"tests: Accomodate some apache2 quirks [1/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5e86891bd8c5509bbf7e0f4828d1610f8007557"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:43:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:20:46.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [2/7]","message":"tests: Accomodate some apache2 quirks [2/7]\n\nApache2 is willing to drop the bytes on the floor on behalf of clients\nif we don\u0027t 206.\n\nChange-Id: Ice38af6427fb6f30e07159a7771f07415f4d223b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543ab8ca1032d8f84d8c009c13447824282d408b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543ab8ca1032d8f84d8c009c13447824282d408b"}]},"branch":"refs/heads/master"},"f19004eb2a2928a34ff31def79b9dbe282bb86d1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-02-10 22:18:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/943355/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/943355/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/943355/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/943355/4"}}},"commit":{"parents":[{"commit":"eebe3ef7eec981e78973750336536736348fc076","subject":"tests: Accomodate some apache2 quirks [1/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eebe3ef7eec981e78973750336536736348fc076"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:43:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-10 22:17:48.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [2/7]","message":"tests: Accomodate some apache2 quirks [2/7]\n\nApache2 is willing to drop the bytes on the floor on behalf of clients\nif we don\u0027t 206.\n\nChange-Id: Ice38af6427fb6f30e07159a7771f07415f4d223b\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f19004eb2a2928a34ff31def79b9dbe282bb86d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f19004eb2a2928a34ff31def79b9dbe282bb86d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
