)]}'
{"id":"openstack%2Fswift~943356","triplet_id":"openstack%2Fswift~master~I0e8f8475148bf4c828b0f4f1088c16fb7df83802","project":"openstack/swift","branch":"master","topic":"https-dsvm","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-05 01:45:07.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0e8f8475148bf4c828b0f4f1088c16fb7df83802","subject":"tests: Accomodate some apache2 quirks [3/7]","status":"NEW","created":"2025-03-05 00:40:23.000000000","updated":"2026-02-10 23:17:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0bfb7474c144610525f84d5b8bc0fbb433be1a4","_number":943356,"virtual_id_number":943356,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-10 23:17:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 01:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b433887686ffc60bd039c7da1ef26eee92f9247e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 00:40:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5add15e3edd7858b6f42c7640c77a714e9475be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 01:45:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0520230548c94dd7bc93e1cb7b3def5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2bdc7e27a7420f8ad86205852b7284 : SUCCESS in 8m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f86853562e6a4aed9447902981eaee7e : SUCCESS in 50m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df77e4a95223445a97aea1869f1ff00f : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b898803b3e34adf997fee65bd2e218a : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/660b5a5bbf664e6b82b2f0785855548f : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8dcdaf13098d49a38da102dc71dbe237 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2a09a8ca14df46a1bf0446c97517a348 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/44c0869475bf483ba04db0f2491d81e6 : SUCCESS in 15m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f8f2174962942a184774e4410db3718 : SUCCESS in 54m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/946792e5322a44cd98684af7dd9ddded : SUCCESS in 54m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f8c7ad0edc4392bd578b8aaf0fc0ec : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac41443ba4324a1688ce7e59b14fb9c2 : SUCCESS in 26m 00s\n\nWarning:\n  Comments left for invalid file test/functional/test_dlo.py","accounts_in_message":[],"_revision_number":1},{"id":"136355c2e55cbf1551455814d6f35e8c2da7ed74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 03:29:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cade4b2c51594cf0ae713b3e5cace275\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dfae5b87c834da298d071d4eb0712b0 : SUCCESS in 2h 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b664e52d0ec2ad34d7b0c2971da1eaec291a8296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 21:50:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d4b7881ba7d46c3198e5cf732ce6cc3d9fbb5479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:50:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e5228fce534ae1ba873c710d368526\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317e1a9e5c40465cb746d056aacfb8c5 : SUCCESS in 8m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b2ebc1b30b348eba5a5d0346c0084cd : SUCCESS in 43m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25ed0ee79b1648b78818aeb3c3b731e5 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6185d4dc738c4927a6501b45a4332389 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c69ee267af334e9485dd619673f1fc3c : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/67fa18a4417b41b2809482bec1ab3746 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/66e2cb654f9248938ff1d1eef18b7d68 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5a6bf0d76c0a4d58889fad534c2f26db : SUCCESS in 15m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20c79d441df54b44b565da7bc57f43ab : SUCCESS in 34m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f0de7f2740f4de3bc329d5156f8f5eb : SUCCESS in 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07987b5aa2d24416845016df0afafef5 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f982d53c576d416f9ab8cdeedeb1af10 : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1ea4031137cf3d7592cde08adad3d500470223fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:18:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1f09b21e3d4b6c907ed4a0305a3b55\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7970802711e411180d80e63fb4f952a : SUCCESS in 1h 24m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c9fc3c83e4812c961180a6114919e0b5b8e950d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 22:28:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"55b1af9db859d17dcbdb7b4b76af8578703c0195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:31:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aef465a84a14471a6faf2f591d976b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558d57f978904dbe91c33563f5becd50 : SUCCESS in 9m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e24a3db6f71c4693b3a6f8671a092486 : SUCCESS in 56m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/015cdcfbbf4b44e68d62c1f45a215559 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6720097784664dc3826927660d84b369 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d471dfacf0a4d0fb96765f7547dc553 : SUCCESS in 22m 35s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ef332d86ecaa4f73b7bbdb971363841c : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b0e0d0cf27bb4cb08446aeea16634f1c : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e4efb1794ff412bb0f749f13a609823 : SUCCESS in 16m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d78622c46dbc4deb96d7bd55768b764e : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b44e58997f9342e4a2444988115bca01 : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf247436165945d49634cf7227c8bb00 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/547862466f8c46d5bf08c18c4b7cc414 : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6738809ccff16476a66935f99f77ac9e750811bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:45:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc6d11c2c324df59e1c17f0a602b819\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8dff039b42045c8b199a57f2053f596 : SUCCESS in 2h 14m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9e17c50df52cb62a2ab5bbde146d096c1b4cef9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-10 22:18:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1eb2bacb3415a37c8d9cf74a011135e251b35a05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:00:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465c5a08881f4452abb642fff20ef849\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/217a858803b648a5b7788830a22e02a0 : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":4},{"id":"c0bfb7474c144610525f84d5b8bc0fbb433be1a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:17:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06405449f0fd42cdb53440748e33da75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f77271f716a47c4842b4168309d91ed : SUCCESS in 7m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/413f3dff668d465c8886357aff1805fd : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d93db34d42b14c978992ab36f81690f1 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0bcf99831ac4645abe80f974abc4715 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/119156722b9b431ea7ecfd228ec369a5 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/06a01f8a68a94d43b7fd27d9287f3d34 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e46ad652df93492fbc05cb32355c1ac3 : SUCCESS in 9m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50427c0fd1a744e9a72bebe5bdeb3ffa : SUCCESS in 35m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eeb686fa17aa476c83b3dbbf04276e87 : SUCCESS in 52m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737cdb92da33409b86c66733f1009aa4 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b3ae12360814976b453918a6e966b29 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"786d2747629dce124047f29292f711854e47a5cd","revisions":{"45bb40994217a9b16a80f99f5dd43ac66252b633":{"kind":"REWORK","_number":1,"created":"2025-03-05 00:40:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/943356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/943356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/943356/1"}}},"commit":{"parents":[{"commit":"2b58014b8a39a73d7a37228e81f080d0fec60b00","subject":"tests: Accomodate some apache2 quirks [2/4]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b58014b8a39a73d7a37228e81f080d0fec60b00"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:44:25.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 00:38:20.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [3/4]","message":"tests: Accomodate some apache2 quirks [3/4]\n\nApache2 will collapse leading slashes.\n\nChange-Id: I0e8f8475148bf4c828b0f4f1088c16fb7df83802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45bb40994217a9b16a80f99f5dd43ac66252b633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45bb40994217a9b16a80f99f5dd43ac66252b633"}]},"branch":"refs/heads/master"},"629d77e8ddf256a4261afdcefc8c06d6f763e312":{"kind":"REWORK","_number":2,"created":"2025-03-05 21:50:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/943356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/943356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/943356/2"}}},"commit":{"parents":[{"commit":"d66cba7a7253d18497981808fbe96100a760a5e8","subject":"tests: Accomodate some apache2 quirks [2/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d66cba7a7253d18497981808fbe96100a760a5e8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:44:25.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 21:49:06.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [3/7]","message":"tests: Accomodate some apache2 quirks [3/7]\n\nApache2 will collapse leading slashes.\n\nChange-Id: I0e8f8475148bf4c828b0f4f1088c16fb7df83802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/629d77e8ddf256a4261afdcefc8c06d6f763e312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/629d77e8ddf256a4261afdcefc8c06d6f763e312"}]},"branch":"refs/heads/master"},"f237fd9f2c6de12a694ead70d0ab2505e5ff31ae":{"kind":"REWORK","_number":3,"created":"2025-03-06 22:28:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/943356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/943356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/943356/3"}}},"commit":{"parents":[{"commit":"543ab8ca1032d8f84d8c009c13447824282d408b","subject":"tests: Accomodate some apache2 quirks [2/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543ab8ca1032d8f84d8c009c13447824282d408b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:44:25.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:21:17.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [3/7]","message":"tests: Accomodate some apache2 quirks [3/7]\n\nApache2 will collapse leading slashes.\n\nChange-Id: I0e8f8475148bf4c828b0f4f1088c16fb7df83802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f237fd9f2c6de12a694ead70d0ab2505e5ff31ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f237fd9f2c6de12a694ead70d0ab2505e5ff31ae"}]},"branch":"refs/heads/master"},"786d2747629dce124047f29292f711854e47a5cd":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-02-10 22:18:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/943356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/943356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/943356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/943356/4"}}},"commit":{"parents":[{"commit":"f19004eb2a2928a34ff31def79b9dbe282bb86d1","subject":"tests: Accomodate some apache2 quirks [2/7]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f19004eb2a2928a34ff31def79b9dbe282bb86d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 23:44:25.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-10 22:17:48.000000000","tz":-480},"subject":"tests: Accomodate some apache2 quirks [3/7]","message":"tests: Accomodate some apache2 quirks [3/7]\n\nApache2 will collapse leading slashes.\n\nChange-Id: I0e8f8475148bf4c828b0f4f1088c16fb7df83802\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/786d2747629dce124047f29292f711854e47a5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/786d2747629dce124047f29292f711854e47a5cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
