)]}'
{"id":"openstack%2Fswift~943491","triplet_id":"openstack%2Fswift~master~Ibd3d5f83fcd31d78447a894ff4426e3e1b62e5a8","project":"openstack/swift","branch":"master","topic":"drop-py2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-06 11:25:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd3d5f83fcd31d78447a894ff4426e3e1b62e5a8","subject":"Remove __future__ imports","status":"MERGED","created":"2025-03-05 20:50:37.000000000","updated":"2025-03-06 11:26:40.000000000","submitted":"2025-03-06 11:25:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943491-drop-py2","meta_rev_id":"1c64f5d80a3b82ceb17961cb3f53a6b3b2e31e3a","_number":943491,"virtual_id_number":943491,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 11:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-06 09:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-06 09:07:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 21:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 22:07:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 09:07:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"8dc04d5c404d6ae0058d7b2cf16663661aa188ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 20:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ff0f4adc06e09d6aa532c2790829569a4098bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 21:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b2846549c594224bf052a9be838a900\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4c1f37b28d5477d8433865efbaa96a5 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"29bbb2ceae925ca2289a0e81fe2955ffb1d4e35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:07:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e230fa48d45a4d20b530106804bc875f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912a54f8c8014211b6a805223323bbb7 : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f09cf3d651b4414c9c42b3e429dd371c : SUCCESS in 47m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3491b98ad5ac45d9bb72d6ca0176aea7 : SUCCESS in 49m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20f64602efaf4a808f834db6c1f4725b : SUCCESS in 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d96b3eea0a874e5d8db94c242ac63e24 : SUCCESS in 51m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1d99c063a2a40f68bf7f4b95273ded1 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0ca5c9fc10541529a8a5ff191dd5908 : SUCCESS in 13m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03cfc52a5581432687a0fd652f2fb906 : SUCCESS in 14m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94a69ed2a5b94134b527c8e1eff32e3a : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53d2d858d549411cac93680aa6e7b8b1 : SUCCESS in 10m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e0cc4f2de6e48628f0a217619c3f853 : SUCCESS in 11m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca7443efd121414f81e035bfaff5d5f9 : SUCCESS in 28m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d389f921b1924ac2ab2e41e630bff50a : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/83728cf824254a25b4669dd0bb42f4b3 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bbd6967834084da1ba27bff422a78c3c : SUCCESS in 18m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da1c62045ac143a1af506f5a75852a6a : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3e71921973a4de19324a59de71ea6df : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bbd6fbd93cf54c0c852d755295aeab93 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6e8bf7e8e3049b3bf7cced517ed7249 : SUCCESS in 51m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4e6cd57c8c14d0abdec0c291259aa7e : SUCCESS in 49m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ab074ec73de42c9929eeaf9da578d08 : SUCCESS in 36m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bb4976ad96147aba0d42046167e65e5 : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6dd625a4f01440cb8b2b7ee1aebff82 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa16a4c37cca4eb8a4ae4022abed9ae5 : SUCCESS in 22m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5639dd1c3ad84cc7ad40235cdcb5f788 : SUCCESS in 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef8b1420d3a645f3bfe32b0970ea32ea : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a9c7e43f418eb3a09174b68aae960292db680866","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 09:07:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9399c8c2a678a2f2c2a3ee85049154d10f22a3d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:08:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b77e126786d612197b8b14ec83e87a25b6aee9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:25:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e66a42f8cd423697a7d0d465f763fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175c6508af98450eaa8e5d8c18944d9a : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/752bba9492ca4282af74704023c8989a : SUCCESS in 1h 09m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6f1a7ecaa474ef1b52fa0b8a711d85c : SUCCESS in 1h 08m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/851101dba74a48c0a6d3c167a0f65a6f : SUCCESS in 1h 06m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/681e7230d25f4cbcbf29f9e3435138cd : SUCCESS in 1h 14m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bbff5274e7ce4f9bb22b8b98936cc5a4 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59af5de961fc4e12807490db755a62c2 : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b0edce9b338462ab87b7b851c8fa500 : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d55f5d4c4eb241949d01667029a5b074 : SUCCESS in 26m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f47227fe26c43babffc0a2e6616bd1d : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/074608d1ab9f46b1a9d468bb8d1903f1 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a2dd0531b58043a4b2f7f308435723c5 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68c49a4b2e684334ba72c50ac54109f7 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46f4524339d248e1b21f31a4b4bbbdb1 : SUCCESS in 9m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c2d8a8903c44533905261c4ab53fcb4 : SUCCESS in 55m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e818b70088ab42fc926ca7b418b96f24 : SUCCESS in 34m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d22ceb505b54461fbbfe30008d75a9d4 : SUCCESS in 34m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ab9309b35374522ba7f4dee81703fdf : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7e39f97b37420b9f4148c4f8ed54a6 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d4555b947c849d39a4ea97289fb0e4e : SUCCESS in 16m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ef1570114d84dc198a1a073b222f29b : SUCCESS in 49m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/176be8e91f364b0587064f95a5844369 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"05878e0bac79e75adfdbb1613474f8d5a70ec441","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:25:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c64f5d80a3b82ceb17961cb3f53a6b3b2e31e3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:26:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edae773a9bf543219a24ce5e993679bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75fdeae393394ccbb08f01dfbf0ffa42 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/65d74377e37f4123a71ff3b7e86cf222 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bae8cd0db8a0ab9021355f99b06d5fcc574d588e","revisions":{"bae8cd0db8a0ab9021355f99b06d5fcc574d588e":{"kind":"REWORK","_number":1,"created":"2025-03-05 20:50:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/943491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/943491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/943491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/943491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/943491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/943491/1"}}},"commit":{"parents":[{"commit":"0a8ecbc554b811d35713a9486774ca93251d0c9f","subject":"py3: Fix unicode-header-name handling in bufferedhttp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 20:50:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 20:50:21.000000000","tz":-480},"subject":"Remove __future__ imports","message":"Remove __future__ imports\n\nChange-Id: Ibd3d5f83fcd31d78447a894ff4426e3e1b62e5a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae8cd0db8a0ab9021355f99b06d5fcc574d588e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae8cd0db8a0ab9021355f99b06d5fcc574d588e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
