)]}'
{"id":"openstack%2Fswift~943492","triplet_id":"openstack%2Fswift~master~I48faabc5b59f94a47dfeb871f13ddbcbaa20d349","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-06 12:22:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48faabc5b59f94a47dfeb871f13ddbcbaa20d349","subject":"tests: Reference test.functional.normalized_urls directly","status":"MERGED","created":"2025-03-05 20:57:59.000000000","updated":"2025-03-06 12:24:31.000000000","submitted":"2025-03-06 12:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943492","meta_rev_id":"57588199a0b4dfe933fa909bbcdff85ebd4ece4c","_number":943492,"virtual_id_number":943492,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 12:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-06 09:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-06 09:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 21:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 22:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 09:16:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"57e09c02fa033cae441df6245d6e395d2f54b9ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 20:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5561458f5d05d71e4abbb2ffedee5635edeb056c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 21:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ecce84f8ebe41b9be184496e29cd2bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3bdb71b1817144f180c48d412afa28c2 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"737ad309333673b364bed906ce1723962e00f508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:03:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9740b1bf7d07498abbb1661ee6910e75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2874801767400681c974866e9e161b : SUCCESS in 5m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/188b2a32454c410ba757dc94228c4fc3 : SUCCESS in 1h 02m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b86706598049443995d8587853682097 : SUCCESS in 23m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6c0b57f3c664b96a0da0bb13b362fbf : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/587f3391fd1c4fbca53c197fecc43658 : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc7fea5732f74bf8a9752765db5f7cd4 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2def24670f9047b19b4ada9d75959aa6 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bdef8ec374c4b39968fb92312952d2a : SUCCESS in 14m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a3413f25b0d49499d0e12ab37b3b88f : SUCCESS in 33m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d371004c8bcb4287860fd2de8f8794b4 : SUCCESS in 44m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0306e247bb417eb768b3c9dc0279f1 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f0c911cd503426ca56c360d25508f77 : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e0b20323e819788a7a0fea32c6e5ab16c666aea4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 09:16:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39938078029d37567e637612a8d96d511c917554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:19:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd7494fcabc330fc270145ea62be3a6211165efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:22:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac77349586a46acbcba92514ff931ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90592785f4fe4e208b86338faf9998fa : SUCCESS in 8m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/584ffd0f9e8d46b191b17f27796f5b6d : SUCCESS in 57m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9049f52bf548e2874e1adc2d7babcd : SUCCESS in 19m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da84d9dfd06b457c856e468ff985af22 : SUCCESS in 23m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/65650d8ae3eb42c88b6755167bbfc076 : SUCCESS in 39m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fcef0e8cda8844b59f1c23d409eedf95 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/346df23f6faa4e1c8f87aec70e37163b : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cacf2c1e9fe34e289dd8b819dad38c4c : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8e304b5f06e4b118776a6fcef5ed064 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d9859b09b77434e95ff1da2de2b6ebf : SUCCESS in 12m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa23a0fb0c134a2582600469cd975406 : SUCCESS in 55m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1816634188a84bf598b67aaece580c98 : SUCCESS in 52m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914be6d998d446ce8333e90ed068b5ec : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd471ccb2e034ce7850a2860f4902d3b : SUCCESS in 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"501951174cf803d2e602d40ec1be8be1bc349a68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:22:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"57588199a0b4dfe933fa909bbcdff85ebd4ece4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1931280be654ae98c8d36b29e26a804\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef22d97ce6cd49cf8b107c12e157146e : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/225e5829a72148c88bd9e38d29b29fe1 : FAILURE in 26s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3db28d41f182877cdb8c401673674ab9539b5e19","revisions":{"3db28d41f182877cdb8c401673674ab9539b5e19":{"kind":"REWORK","_number":1,"created":"2025-03-05 20:57:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/943492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/943492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/943492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/943492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/943492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/943492/1"}}},"commit":{"parents":[{"commit":"0a8ecbc554b811d35713a9486774ca93251d0c9f","subject":"py3: Fix unicode-header-name handling in bufferedhttp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 18:10:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 20:57:05.000000000","tz":-480},"subject":"tests: Reference test.functional.normalized_urls directly","message":"tests: Reference test.functional.normalized_urls directly\n\n... rather than importing it before it\u0027s been read in setup_package and\nnever getting any updated value.\n\nRelated-Change: I699d664bcef6754aac4ec6877830e745b764342f\nChange-Id: I48faabc5b59f94a47dfeb871f13ddbcbaa20d349\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db28d41f182877cdb8c401673674ab9539b5e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db28d41f182877cdb8c401673674ab9539b5e19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
