)]}'
{"id":"openstack%2Fswift~943493","triplet_id":"openstack%2Fswift~master~Ia0349d991435492ac9d80c36c5b1123b9e310db6","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-06 12:22:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0349d991435492ac9d80c36c5b1123b9e310db6","subject":"tests: Allow configuring normalized_urls \u003d false","status":"MERGED","created":"2025-03-05 20:57:59.000000000","updated":"2025-03-06 12:25:43.000000000","submitted":"2025-03-06 12:22:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943493","meta_rev_id":"c582417deac65d08752ec72b9dbff1b88d69e319","_number":943493,"virtual_id_number":943493,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 12:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-06 09:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-06 09:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 21:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 22:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 09:25:11.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5243ac2e61ba370ab73281580e18e8a42cf4bda8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 20:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f41424154d5ff5f568504e4f546981912800bb77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 21:50:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4f4f20f7af4021943d66432dae4fb3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27b771fa263d46db955ef098aeb1465a : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"75f6c63c64dce1af43ae815f18d76663f793534e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:04:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9dcd80feaf34aa1be0624d7055dc06a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aaa36e59e3542a99095e49b08421ce2 : SUCCESS in 5m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0ee65c4cc92442d99070bab611c6496 : SUCCESS in 1h 00m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce98cd39e7da48a9b8a45254c2311812 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bcce90a58ea4f06bc8de228e62aea93 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b90db4a4e24d496b8c5d46725b027d36 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47a63cdb8dac4d8aa326014da18477b0 : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/02272628842b4d93b525002a792a2775 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b56e051ff95448e7a262708415d30793 : SUCCESS in 12m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a415bcf990b49bc92b18f0cc123d308 : SUCCESS in 35m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f1d8a66898e4c6883a877bc9b838a6d : SUCCESS in 48m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5944b1428a7945cf918d8c71ebc6194f : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6e113c5742b43b3a7c6495a69e3091a : SUCCESS in 21m 42s","accounts_in_message":[],"_revision_number":1},{"id":"77a84bd8f99a9b19420b9f4798a038d0dee3e44d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 09:25:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16eb36643ed6c068ab807dcb36c12ffa1ef28636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:25:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ef72f9057d314634e7685f635adceab9c5c257b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:22:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/698ad7a827c4477e9a23af4bca7ffeda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5429c8cd25d8410da56f6a829dec8ad0 : SUCCESS in 9m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3865c8a9c2b345edb28e2ef91b17e90c : SUCCESS in 43m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf7b824104b24c218bf52d9093358d1b : SUCCESS in 22m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6035e5f16b584d8582aa1a77441a048d : SUCCESS in 16m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73777310dee8408db8e51d66cfea4ea4 : SUCCESS in 20m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66f518c590be4268a9200073f6a13c08 : SUCCESS in 12m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f2162eb8d26643b683e8aa272ae0f0ff : SUCCESS in 14m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cce13797fc92491ca56fced61613bf4c : SUCCESS in 12m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aafb2c2e99ac4f7292b27dc838134cbe : SUCCESS in 6m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/98e645534fbc4de3af1de243c90ee315 : SUCCESS in 6m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3720d6e3ea247cb854de2506c597254 : SUCCESS in 51m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9793d2ec10c64362b67f40672fb980ed : SUCCESS in 47m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f267075860481a8a84939a8281e7be : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e502a7ff14d043da8084e94de51cd018 : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":1},{"id":"afd45470ca6d48494539c5c0b41e9f4e19c87fe6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:22:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c582417deac65d08752ec72b9dbff1b88d69e319","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:25:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb7c0c2d3b844a0826f8d823e333955\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6241db3a92d4d2ba6d47752f2e46f93 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dec81f718f14453ca578206e883cf6d3 : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11422a39575afed7cc90001f27607d97a0ff0d2c","revisions":{"11422a39575afed7cc90001f27607d97a0ff0d2c":{"kind":"REWORK","_number":1,"created":"2025-03-05 20:57:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/943493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/943493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/943493/1"}}},"commit":{"parents":[{"commit":"3db28d41f182877cdb8c401673674ab9539b5e19","subject":"tests: Reference test.functional.normalized_urls directly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3db28d41f182877cdb8c401673674ab9539b5e19"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 18:11:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 20:57:05.000000000","tz":-480},"subject":"tests: Allow configuring normalized_urls \u003d false","message":"tests: Allow configuring normalized_urls \u003d false\n\nPreviously, we weren\u0027t converting it to a boolean, but left it as\na truth-y string.\n\nChange-Id: Ia0349d991435492ac9d80c36c5b1123b9e310db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11422a39575afed7cc90001f27607d97a0ff0d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11422a39575afed7cc90001f27607d97a0ff0d2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
