)]}'
{"id":"openstack%2Fswift~943502","triplet_id":"openstack%2Fswift~master~I9044e49ff30b092a97b75fa7fa133016d17cde37","project":"openstack/swift","branch":"master","topic":"drop-py2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-07 10:57:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9044e49ff30b092a97b75fa7fa133016d17cde37","subject":"Remove dependency on mock","status":"MERGED","created":"2025-03-05 22:16:15.000000000","updated":"2025-03-07 10:59:02.000000000","submitted":"2025-03-07 10:57:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"943502-drop-py2","meta_rev_id":"64370faa9bcbb0aec586585520d980c21b9c1c25","_number":943502,"virtual_id_number":943502,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 10:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-07 09:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-07 09:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 23:21:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 23:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-07 09:47:17.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"0c1eeead28d75f3bcc8954c618a5af0da6f7ecbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 22:16:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01e37775656b4d196bdf8ef36fb7f0a097ae74bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:21:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83fd215e34054bb3ae596fffb7011484\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f73659271c7414081706ac00542a4d2 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5ebf41226db6af06cdae587ef013052f20dcafc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:31:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7de3fd1713224890a0edbf483045f2d1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e3e72b3e14314814a89770b217ba4b8c : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341714e60eb249dd8d345b0669d660e6 : SUCCESS in 10m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b9b99ed0a9a4faca9905f88fef18177 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4419fda1d254f62b0f9866382ce4f86 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/65d474ef7d4d470b9c4beb311b168a39 : SUCCESS in 1h 00m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6a5332c8db54c528d186e5bd05d3574 : SUCCESS in 41m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ed4ede422ac46baa9251dc79fd03243 : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95819e441c7442baba62ba072d57ea94 : SUCCESS in 1h 05m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f15d06bbf9d64971b0e5868136cbee37 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fff455df16554242b49ec4380a407c74 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aff42dac3ce043d4a1a943b52b3a84e5 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b6abc3039394ae8b5a78459838f11bd : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/beabe0a380c34bef94043ffcfc0b615d : FAILURE in 4m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97aa60a883934cb09da023cb761b03d0 : FAILURE in 4m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ceb59ef8a244646a689a4c4c3f2114e : FAILURE in 6m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/15771b2bdf2c48db9f6cf8e180150e9e : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1190626de57400cb5cb192e20631d75 : SUCCESS in 12m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77f49a5fa31946b49ff7ad07d34023c5 : SUCCESS in 13m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3865aa2ce635447e9aa67e8446f3e8bc : SUCCESS in 4m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c6bc369438547a588f223b56bd0465e : SUCCESS in 9m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/849e35878f934378b2e20ba7a9fc26c2 : SUCCESS in 4m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e76b4e5938db493e93adb55c91e6eabc : SUCCESS in 36m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f5e4d9e1fc04aa7beddbb1074310d05 : SUCCESS in 28m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6bbb600ecc5545c9937e350d97a7efe9 : SUCCESS in 24m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5880f9f06b64fef848abf252ba08d96 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002bdb630ac64bada359dd48dd4be27a : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c82cca87b2c24a05aa98e246e8d2992e : SUCCESS in 17m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/957f20ec70ed434abb2e30f35f28d814 : SUCCESS in 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ce9431989634f809d34b9613efdd5fc : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4987e19234973273f2f149ca030a12d2ed66fb7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 18:38:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cea357d379814b0a08b085cb1c2732d5b34be8c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 19:45:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92df70658a7146ae9a2c68870cbca061\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f97a307657a0452c894437c82177164d : FAILURE in 1h 04m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ac248ddf8911c41f673cfd0bf7c0e557e3556e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 20:10:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df89a75d500643b4ac931c372692f490\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b86572a2cce646939e42eb8b772d3215 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0641155e7f6f4586bbce21dcfe8a0053 : SUCCESS in 10m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/440ae62365a14bb98d88bae8d8c9b923 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/866eb6cee6dd4766bb50424626bc0311 : SUCCESS in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/b89ce7be2e2e465aa592cb45c6100dbc : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e859b23e89c74e09b995c67510583a48 : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/149596346a16437e8f2bea6a705fde48 : SUCCESS in 52m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87a12c2d9fea42eea1f79ba1340d49bd : SUCCESS in 1h 07m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebf710f97f3e4b95a52fae1af42327d4 : SUCCESS in 26m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada5b1cac85f494bbdba13bdbd68766d : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd7f4ff9eb3c4fbea1391b562a19f1cd : SUCCESS in 22m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f5ea5ce3de148198fe31a248df0cf6d : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c2da3a276b14de4ae24a814e0055875 : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a8b6a0cb8b894ead9ace8f1c8024c057 : SUCCESS in 10m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5f964e60dddb44a9bcbcfdf459cacd6e : FAILURE in 28m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/edf8db66521f4a959da742c2fb4856d3 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/922398ae306f4400b6e6cd8ba166cf8b : SUCCESS in 17m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1619e636c7d44b8e9bd56cbf77bc5a4d : SUCCESS in 13m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2046c759351747bea61a180331d79023 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfec009e609940998b9de22c785102a6 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/782a1fbaabef4b8ab231e8d79d544f25 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d71b14295b44762a4b94d5e7dc350c0 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e59fae397526452685a381333c4cf4fc : SUCCESS in 44m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0af89fc0292541a19611fcc9795db8d0 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2402b44ec2ee44e3a4d40c0dbba44625 : SUCCESS in 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cafa1144f476494e842792649b2d6d83 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/466e503b048d450c99460ee16837cb1f : SUCCESS in 21m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf370f8cd5ee4549990047466bd5ef2d : SUCCESS in 58m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f06a22190cc46c0b341627fdafb0b4b : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":2},{"id":"cefe6dcb03af3a21f0389f7dd192eb331bf27af0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 20:17:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d5cdfc9e24e3ddc22099f1d01ca31e184408fc8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 21:00:03.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9015e605e104961b20306a6b372a40a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d5d578ceb7b4467b1085e731af259b1 : SUCCESS in 38m 28s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/10955519530d42a28b40e7ddef533eb2 : SUCCESS in 40m 22s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d0cfac15b66466cb188ef61626f4367 : SUCCESS in 37m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2e0fba5ef9a4b2bacbcb1181618aa7b : SUCCESS in 21m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0f10a31d82ee43b486be4c3d7a0a8239 : SUCCESS in 15m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4820d2b915df4a5baf6561c877baf0ef : SUCCESS in 20m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cd7f548dbf62429e9ee0e3e1bb2fc927 : FAILURE in 21m 15s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/378a206cae7344feb547927fadeab795 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8d4c9b67567449b8de91208a6072cb8 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1942b6f938a4add8f1b183a5c3748b1 : SUCCESS in 20m 19s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/faa5469e776243e4a31b9d55c71bd4af : SUCCESS in 16m 14s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/ff3654cf48374ed79cf5243740b5e87c : SUCCESS in 15m 27s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/b82ef3f8cc704b349731fbe983cbc4f6 : SUCCESS in 24m 38s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/0221c7e2e7b74ea883403861773575c0 : SUCCESS in 15m 49s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/c44794a9e8a0423caed486e5d70a3257 : SUCCESS in 15m 39s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/462cd104054944879a6cc262a2ba0d66 : SUCCESS in 24m 36s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/d6a6bef4def344989f735f54d662c68f : SUCCESS in 17m 39s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/8611275a46364144a1e9aafb54c6a32b : SUCCESS in 15m 33s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a5303a4cf8ca4348b9371560e001ceb7 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6ddba6453834dfe4fa7512d4e7f02b88efa4d998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 21:09:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9062e2f18fb1a7e49c3ce3a908f8e524c2ae8678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:21:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/715294c4dfaa4a1ba6fa0369089921c5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/536a9a1120024a75be0c5730a197fa85 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001b28bb010e45bab3c03a7ac29ea82a : SUCCESS in 8m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e471337db8f4f2d80375333cc0791d5 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26fa29880d1947c4b048ff92a8d9526e : SUCCESS in 6m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/724e7111095d4f88a6f3ba97a84afd50 : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/079892a4fc6d46e0a1f811c2d341bcff : SUCCESS in 1h 06m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6736399ae194c12bb8527b98b441d5c : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/159b53974c5040568702b85c8fe79cd7 : SUCCESS in 1h 03m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9a0b987e2544618a98dd63d211f510e : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/138dbdf2ccb843b7b871c72f0dc4b4d0 : SUCCESS in 24m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d93ab2b64fe74198b492944106a665dd : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6bd5ff2af6f248798a74a9590e1c27c4 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee4a4d3e6fcc4a9b9cd0d5a6a337fa06 : SUCCESS in 20m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/395434e241164e48addb9afb31ac1c10 : SUCCESS in 22m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/58066fba7aaa46f7b300eb1f1677d2ce : SUCCESS in 33m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4fdbc588483146be9352e07f25083ff0 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d190c195a914acd93a1ee9bda5ca58d : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/231f37415ee64c3c8c44721d4955b792 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/add1915e6f25435498381d222c5e14c9 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d60b081c7f147d986106b91d69570db : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ee028ac478a4bccbc828a4b9915941f : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02770e7fe76a495a980f80f9365d2623 : SUCCESS in 1h 00m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8fca969b514411ea96191a9a0f011c0 : SUCCESS in 50m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21a1f0353ebd438392e168089348afff : SUCCESS in 48m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/860684ea44334802b0dd0407efa7c75b : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56062ef93194c72b3aa6f8aaecd4ed9 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/267ba2fc4c9d4b23bcc9f40be5150f11 : SUCCESS in 17m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b77288420e47f48b415de860076308 : SUCCESS in 56m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbfdda9f1eee478a8a8f086fb4018c6a : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":3},{"id":"2b8719fbd077b7a9d37b29ed1433d597894586f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 22:40:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47af35043bef4e958d1868c776daf541\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a1f09d8f33841edbc725f1f8ca64afb : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":3},{"id":"175bd28b168ef9b5a309011261a51804ec54a22d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 09:47:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"df19c16cbbbf540bd109b66522874ece7141bd6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 09:47:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf9c6bd69bdb5034b0e9b52865f459000f0580f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 10:57:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/426a50a34b3944be8483e2bc328278e8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/22da95dc967a4607946d56c03d2498b7 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbab7a5b5364c5c912bad76f3b73c79 : SUCCESS in 10m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cad2dd4ab5d6427fa0b6327e7e432240 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f44508c8e2454709bcd60d40a06263ed : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/eccd25220c5747a0a3dc2d9f897249f0 : SUCCESS in 37m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/333ab3d604954db883dc26c52350baed : SUCCESS in 1h 04m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8dc88a84779341b8825214e3cbd67a9d : SUCCESS in 55m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b81d1a1a2ea4e3a941fdaea57655bb3 : SUCCESS in 55m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5fbe4d9d2efa4bdf85a85d042fe4dbb3 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/474542a36d2b426dad1eff323cd4b750 : SUCCESS in 19m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/317c8c6cd5a14da08ff5c051f9a5eb07 : SUCCESS in 20m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b93d51989d584a12aa70b8431b7243d3 : SUCCESS in 22m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f702a7caa7a64e23a79cb874f4d44a15 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3840127590624401984c1db7e9ba31ab : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b27acedfef3447e84020b2d13763d25 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1e89561dcd743c19619dc0897e49bd8 : SUCCESS in 4m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb7d02fd960c4f95a3b4b2e6e7983c10 : SUCCESS in 4m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/720c03467ac74b0ea4ee06794778c239 : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/153fad0f92fe44b78b31b86f332c7be7 : SUCCESS in 52m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/414a659e44a3475d8e2c40d0a7615a78 : SUCCESS in 35m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c03306000afd409ab316671cd076bef2 : SUCCESS in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ad86c4219d41a48887ca7687e2711b : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8795dd7bab944629fc7aa598f9c512b : SUCCESS in 28m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b299c4141fc4f7c971ba2ada1c67778 : SUCCESS in 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35e685445632445c843d286f4cb6ea95 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1d690c2d8bb1130dfe2e2c094087c954f0c43918","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 10:57:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"64370faa9bcbb0aec586585520d980c21b9c1c25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 10:59:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e966e7ec4daa4315acfe7f4f42f2903c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/934ff718af4c4216b3dc6804deed001b : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7108d632b047748fcba26eb9ad2000 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3bcb83c6c484848aec21ab03b6fc287 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/48645dbb54324879b6fecbd3ec1841a8 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62cb4736eda9460467538e31c726ce5fa6d4625e","revisions":{"f1c4c0fb719a4a8c362b38b1a7635ea533106ce6":{"kind":"REWORK","_number":1,"created":"2025-03-05 22:16:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/943502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/943502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/943502/1"}}},"commit":{"parents":[{"commit":"0a8ecbc554b811d35713a9486774ca93251d0c9f","subject":"py3: Fix unicode-header-name handling in bufferedhttp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 22:16:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 22:16:04.000000000","tz":-480},"subject":"Remove dependency on mock","message":"Remove dependency on mock\n\nChange-Id: I9044e49ff30b092a97b75fa7fa133016d17cde37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c4c0fb719a4a8c362b38b1a7635ea533106ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c4c0fb719a4a8c362b38b1a7635ea533106ce6"}]},"branch":"refs/heads/master"},"fc5774dde1b5efbb0f7857ee386aa7ced8dba6bc":{"kind":"REWORK","_number":2,"created":"2025-03-06 18:38:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/943502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/943502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/943502/2"}}},"commit":{"parents":[{"commit":"0a8ecbc554b811d35713a9486774ca93251d0c9f","subject":"py3: Fix unicode-header-name handling in bufferedhttp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 22:16:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 18:38:27.000000000","tz":-480},"subject":"Remove dependency on mock","message":"Remove dependency on mock\n\nChange-Id: I9044e49ff30b092a97b75fa7fa133016d17cde37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc5774dde1b5efbb0f7857ee386aa7ced8dba6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc5774dde1b5efbb0f7857ee386aa7ced8dba6bc"}]},"branch":"refs/heads/master"},"62cb4736eda9460467538e31c726ce5fa6d4625e":{"kind":"REWORK","_number":3,"created":"2025-03-06 21:09:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/943502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/943502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/943502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/943502/3"}}},"commit":{"parents":[{"commit":"0a8ecbc554b811d35713a9486774ca93251d0c9f","subject":"py3: Fix unicode-header-name handling in bufferedhttp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 22:16:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 20:35:43.000000000","tz":-480},"subject":"Remove dependency on mock","message":"Remove dependency on mock\n\nChange-Id: I9044e49ff30b092a97b75fa7fa133016d17cde37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62cb4736eda9460467538e31c726ce5fa6d4625e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62cb4736eda9460467538e31c726ce5fa6d4625e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
