)]}'
{"id":"openstack%2Fswift~943506","triplet_id":"openstack%2Fswift~master~I1530c7f83f207668a37e3d101c423d499458f675","project":"openstack/swift","branch":"master","topic":"https-dsvm","hashtags":[],"change_id":"I1530c7f83f207668a37e3d101c423d499458f675","subject":"tests: Accomodate devstack-deployed apache2","status":"NEW","created":"2025-03-05 23:39:16.000000000","updated":"2026-02-11 00:06:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"585b14fd4f1f957c83ab0e31bb687aac8a1e26b4","_number":943506,"virtual_id_number":943506,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-10 23:37:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 00:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbfbe8a3bf898a8e7b9006ee9e0282ea637ade6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 23:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66cd5382f034cdad7d09dda8ecd01fca464759ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 00:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/903b101fac304398bd3d1f835e9ec01f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755a9f938dbb465bbdaf6f8fcb41a612 : SUCCESS in 7m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/545017422a1f449eaa6dabd01c18985d : SUCCESS in 45m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55d9e08e4b7149659f175b1e008d938a : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9263466dcef94c568ed6a75f012fbddc : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59210f5ff3b44d2a8a93a7390b500621 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92339e1f8b48493ba2419f91933ecb8f : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8ecbfc62c30c48ed8861016216ba278d : SUCCESS in 17m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d7c1308a5454824a2d63124702b7531 : SUCCESS in 17m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e9e3aadaf514bb3963e01f85e6f5f44 : SUCCESS in 36m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/818c0a3d7d334d4a8510a2f30a295ff0 : SUCCESS in 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7ad74bc09d4a8986ce214748183278 : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33f0a84194d8401fa54cc8a248adb051 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"eea1c9b104c7bf2c57dcc66750161365c8c0e431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 00:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55191b31f86c4a48a8f5830ed6051a43\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8c7c02aeff548f1953507f95b7f38b2 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ad4b1deefb8e88d9311d475791ddfc3430b86b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 22:28:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab4139fb3fbea47eebf12442441df405b2a261d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:32:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc776b34a4a64ff386893af674f4465b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9689b9c7e524e6ea46a50a73a6f99c5 : SUCCESS in 7m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23750275a1e44fbe8272acfd7c500da7 : SUCCESS in 42m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c32b7b3e8ada4070820896e7a1fd8000 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1380ca5ca08643128ea916d6de85cf4d : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1b3a8414b0e4b9695ab8a3e7cce82b1 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3956b8cb4a384be0b5fb0613b03d96af : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f112d5081fe9466d8fc360174cc865ee : SUCCESS in 7m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2ee46774f4140159d8b4979e7e4b265 : SUCCESS in 15m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3972e2f0d8bb4825b59fcebdef84a639 : SUCCESS in 34m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95c14a6d68e24996a16bb2c9fd4a8e96 : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd55a4d3bac48979be63fb6421ba256 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/198df2f15dd74d29a6642a74e6e3110f : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"2a93c143306d0e30b824d8075252e552e7138906","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:48:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afc1608dd0542bd93b62c549a029739\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbeefc9d38b24bb49957e92c735719cc : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5de5053c5f7ab08f7409f0ea1cab20dbc87901f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-10 22:18:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"66e84e37091c18b208197c9c74452430d44e9613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-10 22:40:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e64a12a6b693dd2a557a93ac9dadbb5dcf12f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:37:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f97585cb4d4ddb9997c42c8d070b82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad20ffb2b44431198f9cecda3feb67a : SUCCESS in 8m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7758529e8660426990e105307590fe15 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18451b09e3b241329e08a9ea451c81d0 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c57cc7cd04e0410f84cd4268c3e6f200 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5447ef65cc3348c9aac81cc7dd01f2e6 : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f42aa205d9044509818b85a8650acc48 : SUCCESS in 9m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd60e3876f8049dd9229b1249924541c : SUCCESS in 9m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e206743b6e034bf1abd856d1bd1aa4aa : SUCCESS in 22m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7671c7e3828e4dfb8d76d3cc69c60f18 : SUCCESS in 54m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a3ca585d1a4f4a9c64c594b9d2cf5f : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb12a40a71874ee1aa6df8eebfdd6eed : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":4},{"id":"585b14fd4f1f957c83ab0e31bb687aac8a1e26b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 00:06:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a642b4f7efc47b9b1ea74c5639c4054\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7f536473bc5413986044918107f886a : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ecbdc48359f2e41d96afd3ff37d5ac5e31752464","revisions":{"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17":{"kind":"REWORK","_number":1,"created":"2025-03-05 23:39:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/943506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/943506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/943506/1"}}},"commit":{"parents":[{"commit":"e1328a42d27de15dbdfe09fb85123bc9b5c7f2d9","subject":"tests: Disable additional checksums for boto3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1328a42d27de15dbdfe09fb85123bc9b5c7f2d9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:20:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:39:07.000000000","tz":-480},"subject":"tests: Accomodate devstack-deployed apache2","message":"tests: Accomodate devstack-deployed apache2\n\nApache2 (at least, the way devstack configures it) is aggressive about\nonly allowing one request per connection, with a config like\n\n    # Disable KeepAlive to fix bug #1630664 a.k.a the\n    # (\u0027Connection aborted.\u0027, BadStatusLine(\"\u0027\u0027\",)) error\n    KeepAlive Off\n\nJust stop having any expectations about Connection headers with apache2.\n\nChange-Id: I1530c7f83f207668a37e3d101c423d499458f675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]},"branch":"refs/heads/master"},"3a044452a1336733e0988975063abbd51064741a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-06 22:28:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/943506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/943506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/943506/2"}}},"commit":{"parents":[{"commit":"44090d3b3bec4b47f2f9aca544b6af3d131bc739","subject":"tests: Disable additional checksums for boto3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44090d3b3bec4b47f2f9aca544b6af3d131bc739"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:20:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:21:20.000000000","tz":-480},"subject":"tests: Accomodate devstack-deployed apache2","message":"tests: Accomodate devstack-deployed apache2\n\nApache2 (at least, the way devstack configures it) is aggressive about\nonly allowing one request per connection, with a config like\n\n    # Disable KeepAlive to fix bug #1630664 a.k.a the\n    # (\u0027Connection aborted.\u0027, BadStatusLine(\"\u0027\u0027\",)) error\n    KeepAlive Off\n\nJust stop having any expectations about Connection headers with apache2.\n\nChange-Id: I1530c7f83f207668a37e3d101c423d499458f675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a044452a1336733e0988975063abbd51064741a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a044452a1336733e0988975063abbd51064741a"}]},"branch":"refs/heads/master"},"26b02569996b641e8343c8bfc8a4ed316f535681":{"kind":"REWORK","_number":3,"created":"2026-02-10 22:18:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/943506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/943506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/943506/3"}}},"commit":{"parents":[{"commit":"333a0367a2f6af7fd8820c8edad13be5a8067163","subject":"Add proxy_forwarding middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/333a0367a2f6af7fd8820c8edad13be5a8067163"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:20:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-10 22:17:48.000000000","tz":-480},"subject":"tests: Accomodate devstack-deployed apache2","message":"tests: Accomodate devstack-deployed apache2\n\nApache2 (at least, the way devstack configures it) is aggressive about\nonly allowing one request per connection, with a config like\n\n    # Disable KeepAlive to fix bug #1630664 a.k.a the\n    # (\u0027Connection aborted.\u0027, BadStatusLine(\"\u0027\u0027\",)) error\n    KeepAlive Off\n\nJust stop having any expectations about Connection headers with apache2.\n\nChange-Id: I1530c7f83f207668a37e3d101c423d499458f675\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b02569996b641e8343c8bfc8a4ed316f535681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b02569996b641e8343c8bfc8a4ed316f535681"}]},"branch":"refs/heads/master"},"ecbdc48359f2e41d96afd3ff37d5ac5e31752464":{"kind":"REWORK","_number":4,"created":"2026-02-10 22:40:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/943506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/943506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/943506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/943506/4"}}},"commit":{"parents":[{"commit":"333a0367a2f6af7fd8820c8edad13be5a8067163","subject":"Add proxy_forwarding middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/333a0367a2f6af7fd8820c8edad13be5a8067163"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:20:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-10 22:40:05.000000000","tz":-480},"subject":"tests: Accomodate devstack-deployed apache2","message":"tests: Accomodate devstack-deployed apache2\n\nApache2 (at least, the way devstack configures it) is aggressive about\nonly allowing one request per connection, with a config like\n\n    # Disable KeepAlive to fix bug #1630664 a.k.a the\n    # (\u0027Connection aborted.\u0027, BadStatusLine(\"\u0027\u0027\",)) error\n    KeepAlive Off\n\nJust stop having any expectations about Connection headers with apache2.\n\nChange-Id: I1530c7f83f207668a37e3d101c423d499458f675\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecbdc48359f2e41d96afd3ff37d5ac5e31752464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecbdc48359f2e41d96afd3ff37d5ac5e31752464"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
