)]}'
{"id":"openstack%2Fswift~943510","triplet_id":"openstack%2Fswift~master~I0b0bd367f93d4a5529d022f251bba5b99b824a2d","project":"openstack/swift","branch":"master","topic":"inflight-tracking","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-06 10:07:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0b0bd367f93d4a5529d022f251bba5b99b824a2d","subject":"http_protocol: cleanup txn_id error logs","status":"NEW","created":"2025-03-06 00:14:03.000000000","updated":"2025-03-06 20:18:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"274934a858b06460ff5e166d7f7d260aa667372e","_number":943510,"virtual_id_number":943510,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-06 19:57:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-06 18:40:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 01:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 10:07:37.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-03-06 15:22:26.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2025-03-06 18:42:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"b40be3ff2094c1e35bd82f87bd7a99aa1f2c9be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-06 00:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cade08174ea48937afcad8c0b3b89b6965e6955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a850d49b24604de29fdb5321ce27830e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f7e7037e304b31987f90e7b9cf6dff : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/7b7629a846c8489ca57674c4afc514a2 : SUCCESS in 49m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06dfd2d0a4454c8ca8cabba680c24a10 : SUCCESS in 44m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d1e7f2880bb4554b609a0bcfac3107f : SUCCESS in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f11cd0e709e49d68592b32c23f26177 : SUCCESS in 49m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e5be5764aba475a86af72bdea690a60 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdfcd9215f9342a881454b3c30a2bd02 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05c800dc3abb4ab094cef4dc5d901997 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03bcdf7fbbc446229b1a4e7b034fe1bf : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e151c93422a54ff989f67afc9d091398 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb51a8534f02468db8d06bcb63383ca9 : SUCCESS in 16m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db5de0a94dba44b6a0060fd29b6c2282 : SUCCESS in 31m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1ba59283993140b8914befadd6511143 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c0a1639c910a42c8b4d124479597f2c1 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/96c7723da1fb4373b40a23fbd3e22c33 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86c77b881ef24140b0a687dec1650576 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51bc72b58e874456948f2a9024cc692c : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0f9dc8dff5254468891e54a49e59481a : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35d4b9d3849b46c08b6b3304c9a10ba2 : SUCCESS in 51m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/582200de4a894eb19675ea69881439ee : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/777ab13dfc694df4ba633cd2c2c0e7d3 : SUCCESS in 45m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f730970f86714c9eb1e6d00a415a53e4 : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3734dafe7c594358829ba1e0ebf5513d : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4576b9b4f9794337847e80fab1652cf9 : SUCCESS in 20m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/313817add2494a62ba027f10f983b209 : SUCCESS in 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9ab6b9806ac49b0a6a452c917ee4f7f : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":1},{"id":"077757ae4acf482907a055fe4037e6b4c45cc86d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:20:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f6f41d96d7449efb43c9d9426cccd5c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbd64ed3538f4393a0f88b3f2abda2ab : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"19bdc0b4f27ab832cf771218d0ab4f8a8b10c5b0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 10:07:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b823238a9b5928a18cc42e5d839bec1e9c0d9386","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 10:08:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b58e4bd139deb669bdbea314486d650d6c967f2d","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-06 15:22:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c2cf55f5d136b6e1385dbefdfd391c4d134fe29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 18:40:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2d19c206eb75123f59115590dd30be0f290f17ec","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 18:42:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d44feace6f75164ddfedf0ced9f7d477abe071c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 19:57:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18c243ef3d6419ab58634ef0b1bdb56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26093d3105044cd39bd5a9713ad2ae9d : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/56992a9806a145ab8fd1cd31fc3b47e9 : SUCCESS in 1h 10m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dce1c5672faa487b8efe910990fb80ad : SUCCESS in 1h 00m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6dd368899464bdc848d1e9f73fd27f3 : SUCCESS in 46m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af20dace2334359a0e13d3a3bbfe08b : SUCCESS in 1h 10m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aacc311e1581489d94021efa5b0083ed : SUCCESS in 24m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4b0e16342eb4432bf7842ad64dabae0 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1b4a327dbce4f5aa6f0c748d1611085 : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e79645d3b8b442338c864bda78675787 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7736df55f8ec4a86a781ab59c58be75d : SUCCESS in 19m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db8ba560b09e4237b7ad1c3bc4fbc55d : SUCCESS in 23m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7534e3fde5a243bdbd63ea55b92417d4 : SUCCESS in 21m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e782cc0a9b94684804bc0addbc9b60f : SUCCESS in 12m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3889e526ff1f40eab61a0d74758a7c9e : SUCCESS in 14m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29632134dedf40ff8f20675887287f5e : SUCCESS in 14m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/068f4dafbf57427e8670cace8b28808e : SUCCESS in 9m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c443b2d61634a0e8f6e0104ab102688 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e3dd8d95f264d5c8ae6b414a57a6402 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/08847598a32840059e6ad2690bdbcd6b : SUCCESS in 54m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69cf8c9d1d82417a9c932af2aa19f08c : SUCCESS in 38m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4496d4ccc584ffcb0a97c87417d8161 : SUCCESS in 48m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c279e250a841c7bff7c8e390beab0e : SUCCESS in 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163319e8a5f849f28ac911a771cfc5a4 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a53fa33e7a3141d2a8216717f785c23c : SUCCESS in 18m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c21963fd768459eaaa8623e5905aea6 : SUCCESS in 57m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ffd40949fa940a5875e3d4d173a3373 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"274934a858b06460ff5e166d7f7d260aa667372e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 20:18:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b17e6f5b15c04f7da3ce32fe39b7f21b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c69bc09814974a4abe18a1bf1c5ba782 : SUCCESS in 1h 35m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1961e21f2726cd1ccf200603cecf7194f8bbbb3a","revisions":{"250b419dad56e597eaff05817f6545779c023e6f":{"kind":"REWORK","_number":1,"created":"2025-03-06 00:14:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/10/943510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/943510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/943510/1"}}},"commit":{"parents":[{"commit":"c9da7202cff70a2b3d8ad15e79dc83c76f00cb60","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9da7202cff70a2b3d8ad15e79dc83c76f00cb60"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-06 00:13:33.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-06 00:13:33.000000000","tz":-360},"subject":"http_protocol: cleanup txn_id error logs","message":"http_protocol: cleanup txn_id error logs\n\nChange-Id: I0b0bd367f93d4a5529d022f251bba5b99b824a2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250b419dad56e597eaff05817f6545779c023e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250b419dad56e597eaff05817f6545779c023e6f"}]},"branch":"refs/heads/master"},"1961e21f2726cd1ccf200603cecf7194f8bbbb3a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-06 18:40:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/943510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/943510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/943510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/943510/2"}}},"commit":{"parents":[{"commit":"5554f432f5d285c82cd82e40607067cd7a214542","subject":"Merge \"tests: Allow configuring normalized_urls \u003d false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5554f432f5d285c82cd82e40607067cd7a214542"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-06 00:13:33.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 18:40:10.000000000","tz":-480},"subject":"http_protocol: cleanup txn_id error logs","message":"http_protocol: cleanup txn_id error logs\n\nChange-Id: I0b0bd367f93d4a5529d022f251bba5b99b824a2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1961e21f2726cd1ccf200603cecf7194f8bbbb3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1961e21f2726cd1ccf200603cecf7194f8bbbb3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
