)]}'
{"id":"openstack%2Fswift~943511","triplet_id":"openstack%2Fswift~master~Idbeffc6bc3cf31dd4de311aa9ae8f332740ce7a3","project":"openstack/swift","branch":"master","topic":"inflight-tracking","hashtags":[],"change_id":"Idbeffc6bc3cf31dd4de311aa9ae8f332740ce7a3","subject":"wip: emit gauge for inflight_requests","status":"NEW","created":"2025-03-06 00:14:03.000000000","updated":"2025-03-06 01:52:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d38f5ee21ca7af7934d64866b633ea1e933bb284","_number":943511,"virtual_id_number":943511,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-06 01:20:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 01:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b839e273ce4029c9577b0ccb673f6f220359c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-06 00:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f38f584f62dc526662e42ff90c0d52f63648af99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:20:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07d04e3d7da4ac09fbccbd2218c08ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4729097b7d424e25a36d3546296f5394 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/05f85f7b8282447e9791b8ae1ea0c8c3 : SUCCESS in 55m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b18d4cd22fb942a4ae3cd697c72a89f2 : SUCCESS in 52m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf98693fe7af4c6884f89de687b27540 : SUCCESS in 51m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1ad84d78c0f48a4add0eacfd14406f8 : SUCCESS in 1h 04m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b0b0ea8aaec46ac86d463b0985d0515 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cabf1ff33b2746378a9f56db9d68e508 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf913b8dd4d44c37b404328811067fc4 : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c06909f403f64f3098a98b6d74cf8599 : FAILURE in 3m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3a8afc416f4f549594e1b2e714e1eb : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29fc573f7b3644d1819a1a862858a347 : SUCCESS in 18m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b04a015f053244f3ae13710b615ee091 : SUCCESS in 28m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/26e49df7e37543ac97ca3c00f940efa8 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f2b90e244d0d4cb0b4d42e567725b379 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c1df27bfb61146c091c4a8160d071891 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50a0f3bbf1e94c6c93088f9ad481b7f1 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e022e1bb14741c8916ccaa83da480c9 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/086036ca11d94727837fb92572fe3a2f : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/044ed3ae693346a1900650aa51a4e55a : SUCCESS in 57m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6911e55db27b448a817b86536427ba64 : SUCCESS in 44m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0376af9f1e74fc0a287e29daddb28dc : SUCCESS in 50m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d196a9dc5e924fd0bdfc5fe0ba8d96fd : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ecfa2bbc784cc6a9c57bbfc5f2a04d : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2eeca77733074d95879560878e195031 : SUCCESS in 16m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9464b60aad6f4e9592d0cd7f7f8f207a : SUCCESS in 55m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5206bd0a8c28487f98ee38a3acebe512 : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d38f5ee21ca7af7934d64866b633ea1e933bb284","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:52:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b63a8736b54c22a021c312fb0468ce\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e70d582a65824d83a8f5e19ed5942571 : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d809285686929c7c93fa0ad9d121d731b5e9d05","revisions":{"5d809285686929c7c93fa0ad9d121d731b5e9d05":{"kind":"REWORK","_number":1,"created":"2025-03-06 00:14:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/943511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/943511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/943511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/943511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/943511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/943511/1"}}},"commit":{"parents":[{"commit":"250b419dad56e597eaff05817f6545779c023e6f","subject":"http_protocol: cleanup txn_id error logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250b419dad56e597eaff05817f6545779c023e6f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-06 00:13:52.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-06 00:13:52.000000000","tz":-360},"subject":"wip: emit gauge for inflight_requests","message":"wip: emit gauge for inflight_requests\n\nChange-Id: Idbeffc6bc3cf31dd4de311aa9ae8f332740ce7a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d809285686929c7c93fa0ad9d121d731b5e9d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d809285686929c7c93fa0ad9d121d731b5e9d05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
