)]}'
{"id":"openstack%2Fswift~943618","triplet_id":"openstack%2Fswift~master~Ie2fd55c1e7c7e360549932552d904e3cac6e94aa","project":"openstack/swift","branch":"master","topic":"p-systags-3","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-07 18:37:35.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ie2fd55c1e7c7e360549932552d904e3cac6e94aa","subject":"migrate encrypter to use systags callback","status":"ABANDONED","created":"2025-03-06 15:29:54.000000000","updated":"2025-03-11 13:41:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73f172c7264247b0727f6165fe0315b8ee1df060","_number":943618,"virtual_id_number":943618,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-07 19:45:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2025-03-07 18:37:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 16:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aae111a3701792f31b614bd53b00a4c72c1bc338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 15:29:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d752a290a5a024840b356b51207982a6396c2029","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-06 15:30:19.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1506b53951adbd2a3f290bb80fac8e7c818e67e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:56:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43d04f0d29524229a9d1f1f6a1d88343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e217db83ff54b3181ecbc8a3dc2be23 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/f2726007eff049e29bcbeaeb0810fa41 : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a55e2d4eba34aeabdc26dbd06469087 : SUCCESS in 1h 06m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca778dba785a4120bc2675d434850931 : SUCCESS in 50m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff7f081f33184ca79681fa0300376c16 : SUCCESS in 1h 09m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22484a2df5574a9cabd220020bf6c0de : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe8a05bf4a3543acb30a55082366cbb5 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07bf090fbc914c19b3f3dadbd6428253 : SUCCESS in 22m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2a185a0a4cb4d92ab2da7dcf042def1 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c50d736e35914e0eb4b6116e51e435fe : FAILURE in 19m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f777baaacb1241d98f45efba5538b735 : FAILURE in 18m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a50b2b789d147b8ada734fa0f5b3945 : FAILURE in 21m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/254dc7cb5d7a47fd902f0c2927cd8ff2 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/80f3ff7a682f4c8db5223ff55b39e545 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e335ffc5cde462e858f15fa3fccce0a : SUCCESS in 13m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a793668cac8340519fed8d3ddbd43203 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/368f04c8f3214a0fbbdd21a91102f0b1 : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45403efa88da44c0a7c05302987f0781 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af88bfe62aef41f9bf9759bba70c9475 : SUCCESS in 45m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d184850cff2c44e38ec3c2a7e89d2bc9 : SUCCESS in 38m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f688a682c5d642538e3e9bdf9bb99308 : SUCCESS in 35m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfb7ed26869648229eccdce2da2a6fe1 : FAILURE in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec307e96f9249c982d26f9e10b7f92e : FAILURE in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fbad79a124be40fea4f1012aef84f3c4 : SUCCESS in 17m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8ddf0383ed249f4ada8a1b6fe735e85 : SUCCESS in 1h 04m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a19ed5cf06c4b76a7c158ca876f35e4 : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5b45a7b29e814bccd7ec8d24e91b59bb2f466363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 17:50:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d7018b964f146df8dfa4580a1a51717\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e149172a9bb440b7968a773e95b5a4a5 : FAILURE in 2h 18m 22s","accounts_in_message":[],"_revision_number":1},{"id":"408bd2afd525b86e87b89b6bea03bb1f7bbeb655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 18:37:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"401090298b09c360018cab816ed8753211dcca79","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 18:37:35.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"407d7dedca473ed3a19dbb172fcbad9827c6b98c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 19:36:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0399ffc70eb24a0c802d9332891c7afe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21ab8438d8eb415d8487ff4610d6e69e : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":2},{"id":"184f81afa6a6eceb83188b675fa1f1e33fcff04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 19:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69f497192124de5a840a275a55c9da5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ef5b59c66b4490ad0560ecb08cf5ea : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/be73e2b0c70141d091d36e32023d881e : SUCCESS in 56m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a2f609bb5404a8c9d862d2efd8d7b8d : SUCCESS in 1h 00m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b1b89c9a7454204974a5c9765ce4085 : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcbfde342cb9422d96b269a6bcc0f56a : SUCCESS in 53m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f664bfca5803429db7d1b58771bab2a6 : SUCCESS in 24m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/308468181f1a46bcb570dfc4d5760eca : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e815ced6aec4b88905c7a0b18ebed66 : SUCCESS in 26m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c5f023503684249a5151e02950fa6b8 : SUCCESS in 4m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82abbaf50f964918860b57faadd26622 : SUCCESS in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55b851e2f5ce489481d39083984e6501 : SUCCESS in 15m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2713fcba4a6f4cc2ab081e4df2e13f60 : SUCCESS in 24m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/78dc3ae45f694f258b7855b3b4691d7d : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4f5057a88e8141858e1f7d0edd3ce96a : SUCCESS in 15m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dce93a942b884699b3639c8ede18c7dc : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d8925d7944c470c9e949303bb3f35fb : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42e06b0bcccc4139b24796aabbf025e8 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4422c02d223a475ca248c7cd80ca54f3 : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4599e6b64d0045629f976123c7242df9 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86ecb366ae914e819911d8c62f277f52 : SUCCESS in 35m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9c079015e48434aa826d78bb541cbfe : SUCCESS in 44m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c45ab7ec308435da489d09422498080 : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f47562c1cd4d04810750441ed2f736 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b46a10384e3f4f59b7f1ce4f4fbe3f70 : SUCCESS in 23m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9129d2b7711d45a7a915ac4023997683 : SUCCESS in 37m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b46683a67e914ce391ed6aa3f4d9413d : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":2},{"id":"73f172c7264247b0727f6165fe0315b8ee1df060","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-11 13:41:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"857a3dc9e4cbd6095f700034c8be8d5fd9459be1","revisions":{"2fadd9885f079eb7ba2ca4917c00fafa41d21835":{"kind":"REWORK","_number":1,"created":"2025-03-06 15:29:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/943618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/943618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/943618/1"}}},"commit":{"parents":[{"commit":"107fd13efaf7cdac402ba6d799a9a835791827d5","subject":"WIP proxy: Add systags abstraction for etag override","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/107fd13efaf7cdac402ba6d799a9a835791827d5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-06 15:27:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-06 15:29:37.000000000","tz":0},"subject":"WIP migrate encrypter to use systags callback","message":"WIP migrate encrypter to use systags callback\n\nfailing tests\n\nChange-Id: Ie2fd55c1e7c7e360549932552d904e3cac6e94aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fadd9885f079eb7ba2ca4917c00fafa41d21835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fadd9885f079eb7ba2ca4917c00fafa41d21835"}]},"branch":"refs/heads/master"},"857a3dc9e4cbd6095f700034c8be8d5fd9459be1":{"kind":"REWORK","_number":2,"created":"2025-03-07 18:37:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/943618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/943618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/943618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/943618/2"}}},"commit":{"parents":[{"commit":"e64773d499500376d4515684bfa1743b199d1d21","subject":"WIP proxy: Add systags abstraction for etag override","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e64773d499500376d4515684bfa1743b199d1d21"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-06 15:27:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 18:36:21.000000000","tz":0},"subject":"migrate encrypter to use systags callback","message":"migrate encrypter to use systags callback\n\nChange-Id: Ie2fd55c1e7c7e360549932552d904e3cac6e94aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/857a3dc9e4cbd6095f700034c8be8d5fd9459be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/857a3dc9e4cbd6095f700034c8be8d5fd9459be1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
