)]}'
{"id":"openstack%2Fswift~943649","triplet_id":"openstack%2Fswift~master~I1b9d2a3369279a56962b800c27edd9f6127ef0d4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-07 11:06:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b9d2a3369279a56962b800c27edd9f6127ef0d4","subject":"Remove notes about new-style classes","status":"MERGED","created":"2025-03-06 22:48:23.000000000","updated":"2025-03-07 11:07:11.000000000","submitted":"2025-03-07 11:06:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943649","meta_rev_id":"b2d321f109445035b8d1378d6291a26ff35f26f4","_number":943649,"virtual_id_number":943649,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 11:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-07 09:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-07 09:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 00:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-07 09:52:58.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"2ba3142e66a40bebf18acd9e644aff6421ae75c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 22:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e1bf653b54509bfbbe8277e8bb07ea44dbe7e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:50:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8b01cde403e487fa0b785fa0501c634\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb29d69b47a415fb53fcc880fad59cb : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/31c5738897c9416e82bab7a2ced6041e : FAILURE in 35m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80697e56edee419fa3d05f049071bcaa : FAILURE in 34m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afc7fbb5cd05480ea53534c00fa80d18 : FAILURE in 17m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f23a635042a445ce9cb20fbc61be6301 : POST_FAILURE in 17m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f837d1f6a5a4b2db8e39963f512a373 : FAILURE in 6m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c1314e3b52d407c97429841ed92c3f5 : FAILURE in 6m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/197947cd4e494006b444d5ac448cb4c0 : FAILURE in 5m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b459585863542d08efaa132721ee448 : SUCCESS in 3m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/157dc695589d4fa09b938767314eabcc : FAILURE in 28m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f2efef0ef5a45edb2074995794b9b6c : FAILURE in 29m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25afc2bc62b64b389670fbe1b7b14e00 : FAILURE in 49m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f33ae7f26ebd4b7586799728f9c5f15b : FAILURE in 2m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e7bf5a650a4142ad808c53ff6b285701 : FAILURE in 2m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc1972d40e094a7aa8d4d07e91b778a9 : FAILURE in 4m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50dd8d3fc9184c9193cf709836819e21 : FAILURE in 3m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d521f3255ce4c868c8e4f1d28f4e089 : FAILURE in 5m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/02fed1b22ad04bd181fcf97f7d613aed : FAILURE in 15m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/26defbf1f41c4efb80b7d8ab5df00c15 : TIMED_OUT in 2h 00m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fcaa3d4c87044879a8abe7d782f80db : FAILURE in 30m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf4d35e974fc422aa2c3316bb6a156bd : FAILURE in 8m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08a32aa63c1d4a0a9dd16b782cc86db3 : FAILURE in 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4b8c7b842f461cafdb27effd3c7a8b : SUCCESS in 2m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ffb6bbfacca411698d6084dbe4235e2 : TIMED_OUT in 1h 30m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cab8b904030147c1af95df8d8503afc6 : FAILURE in 11m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81e57c33504447c7b911698178b62aae : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ea5f205dec63ab6d7412b6494e6ef291d1d93aa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 01:19:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18d8b379f734d1880e0afb6568f214d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc49354a1fd34ca0a7c96a8703b0d5d3 : FAILURE in 2h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4c155a83b873dec900d402aa2cefea0405606dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-07 03:17:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a36da9131f11bbc2985aafc6692fea839e6ecd7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 04:22:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6ced9f60a94ad6b599821011a11f6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e8218aab874f83830c5a0fb80ab1bd : SUCCESS in 5m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/fa7ac069f0d74451a86baf887f6ebf6d : SUCCESS in 35m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b263bd68a4024c2aaa00dab3b9d5d385 : SUCCESS in 54m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb2376359f9846cabb8074e6af45044d : SUCCESS in 48m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/959c99605b34425183d74ccf88dd8967 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/241b7bf1a5394453bc68eb64cc13b975 : SUCCESS in 9m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9dd3cadfbac4940befe6f4d410b436c : SUCCESS in 9m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c68a133447e44ce9afe9e6a82dff88e4 : SUCCESS in 10m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee35feed124e4ba6a21cb8aa89353bd9 : SUCCESS in 3m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bc781c338134b798c4774e5c9ad42ee : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1493414dbe9649dfb8f01a96406f5fb8 : SUCCESS in 15m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/06435d3b5e2f4a789da3bb272bf55648 : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7d21526646c24394b065e436028587e8 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/72bd5a8e9ae04c5c8ede0cf521ad72f7 : SUCCESS in 13m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fe6530fb7424b46bd4a0684a2e79b09 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f421165bd0574fbaaaa5a70bed716de8 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8612362466e04d7f90ad119f1e25e4aa : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6022b12e8d7413cbbc006fb5f21dca2 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a43a49f163f447c48878a41fb728dd0c : SUCCESS in 55m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c094cf932584ab9ae63702ee2dd9d8d : SUCCESS in 50m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/187235534692456f8028afa5c2853179 : SUCCESS in 51m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b08c487c5c34426a27b5cc90c66eed0 : SUCCESS in 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12749184f94c47c381ea41e107076c90 : SUCCESS in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d86cccbe78584451a053c8bc85407974 : SUCCESS in 24m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f35b4d4df2b4b44a23a9fa1a18acc31 : SUCCESS in 38m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70a8d8b556c748c8bdaad36ab8c484b3 : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":2},{"id":"869bb3f9d482d72980e66e22b3da8c3c95f254c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 04:40:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d25df53352446deb7fa0441fe16c1ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cc0891189a64417bd4f183e9a7e1f7b : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a16550e2117402c6e261584bb668a1c30eef8eac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 09:52:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4a9d388a0e6fb06bae7ffe8b4acd4bac6369a92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 09:53:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f757f8979c6be1defc08b09a20cb696856df7802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 11:06:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e435f4ea34fe4c619b8ba9b83ebc2a33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd003428b8e5460fb81245335ef982d7 : SUCCESS in 4m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/30391f9d7d23408ba5de0958701adb5b : SUCCESS in 27m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a290e28b2c794f5780490fa79f525820 : SUCCESS in 57m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa6cbfe608924668b76b899a7e75266c : SUCCESS in 1h 00m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f713880b7af8400eaad8bc05f6dbd923 : SUCCESS in 41m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d88a55539ac149c081a5b052037de256 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b45292fa3a5402a9dbd582b51ed0fad : SUCCESS in 10m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aadcfd00d3d94fc8a94a68e153e577f5 : SUCCESS in 22m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ca3bc66f3454f01a5cc7351c5797863 : SUCCESS in 21m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f817c70cd7b64787b694ccd4161fc7c4 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ae54d1f851844a1a3d31513413139ce : SUCCESS in 17m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a6a02064d6f646e98b7c4c64e8cfd126 : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d66123ee88a4b028cdff4ade0a72dcb : SUCCESS in 5m 55s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8135c55edd549b68e434fb80d41c08b : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6d5ed997f734993ae05a40e59da2234 : SUCCESS in 58m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae37ed3099194802a3b46eac7fc96dab : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b121052ae6924cdabfa484ba13d5f9c1 : SUCCESS in 37m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec47a822b2f84a3b8dcc9b660a297fef : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf397028f39344caaa1858a4677d7587 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12e0205274bb4ea09428679c422f7436 : SUCCESS in 19m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db06de497c5e41da8f8bc3a7abf8437e : SUCCESS in 40m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9704522ebbbd4d728adc900b682472f8 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":2},{"id":"55c793551ea8117655e9743d0bc87d35ce9ed25c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 11:06:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b2d321f109445035b8d1378d6291a26ff35f26f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 11:07:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b7b544d0dc4cd3a4b7133f03efebc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b046151353e547f2845e548b019ee8ab : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/daf76400b15d46668d6142faed807a50 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"932bc110cf4a653941685b617f8dcb1ec2223ca7","revisions":{"5a8796013267b6524ec6ce387b07a0f3b0693054":{"kind":"REWORK","_number":1,"created":"2025-03-06 22:48:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/943649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/943649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/943649/1"}}},"commit":{"parents":[{"commit":"5554f432f5d285c82cd82e40607067cd7a214542","subject":"Merge \"tests: Allow configuring normalized_urls \u003d false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5554f432f5d285c82cd82e40607067cd7a214542"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:48:11.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:48:11.000000000","tz":-480},"subject":"Remove notes about new-style classes","message":"Remove notes about new-style classes\n\nOn py3, everything\u0027s new-style!\n\nChange-Id: I1b9d2a3369279a56962b800c27edd9f6127ef0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a8796013267b6524ec6ce387b07a0f3b0693054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a8796013267b6524ec6ce387b07a0f3b0693054"}]},"branch":"refs/heads/master"},"932bc110cf4a653941685b617f8dcb1ec2223ca7":{"kind":"REWORK","_number":2,"created":"2025-03-07 03:17:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/943649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/943649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/943649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/943649/2"}}},"commit":{"parents":[{"commit":"5554f432f5d285c82cd82e40607067cd7a214542","subject":"Merge \"tests: Allow configuring normalized_urls \u003d false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5554f432f5d285c82cd82e40607067cd7a214542"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:48:11.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-07 03:17:35.000000000","tz":0},"subject":"Remove notes about new-style classes","message":"Remove notes about new-style classes\n\nOn py3, everything\u0027s new-style!\n\nChange-Id: I1b9d2a3369279a56962b800c27edd9f6127ef0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/932bc110cf4a653941685b617f8dcb1ec2223ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/932bc110cf4a653941685b617f8dcb1ec2223ca7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
