)]}'
{"id":"openstack%2Fswift~943683","triplet_id":"openstack%2Fswift~master~Id78ac1cac833f9284946cd364b3f95341afcb235","project":"openstack/swift","branch":"master","topic":"p-fake-dummy-value","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-07 18:33:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id78ac1cac833f9284946cd364b3f95341afcb235","subject":"Fix FakeSwiftCall partial env copy for None values","status":"MERGED","created":"2025-03-07 13:05:07.000000000","updated":"2025-03-07 18:35:20.000000000","submitted":"2025-03-07 18:33:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943683-p-fake-dummy-value","meta_rev_id":"a0d6edf5512f7c19fcb7875ada8dcdc2a30d1911","_number":943683,"virtual_id_number":943683,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 18:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-03-07 14:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-03-07 14:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 13:55:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-07 17:01:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"51ebc343941176c815f7cb0c99c88d0b9c712774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 13:05:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a00967350b8ca7fad68d1d3c891819e8c56cd529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:55:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff59681d1164e328a8dd1afd2c87274\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731ce370901445018b735d750bd8554c : SUCCESS in 5m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c25384e2e03c4aca9e0b2c17fcbca093 : SUCCESS in 33m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63ede89a3d3947af8fd5d88a06afc9ae : SUCCESS in 11m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e661da3c98db409fb3a30580adf7d705 : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f6252994fb4e91b963e2612fd6527b : SUCCESS in 13m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f13352dcd07842f8bf13f83555916dcc : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06af614706804b55866640b38f353f77 : SUCCESS in 18m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aca7ae11061b4a7db58ffe7e4c56cac3 : SUCCESS in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ef611128deb474ea3884c33330082b2 : SUCCESS in 22m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1cbf9ed460024beea5000456dd4ad40b : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a9591c27f7214016bc903c6ef00d6647 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a9cae182cd33413ab42a04627091404b : SUCCESS in 15m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/239837662643414e9ec68937203a2f76 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab9b5fd1530349d79f79aee86824bc75 : SUCCESS in 42m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02039f215ca74268936c449fa88de653 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41331fd86ee9455d8bac7e50ba091750 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fca10b2148e1445bb1b926554b83cd67 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f04937a1c7aeeac980a25cb3819af6e08d3d278","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:04:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77f0bce7b4b411e8caf509b0f3899f7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31a4c0dfa0f4466d93a0265416a71ae8 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ee1cc8a8355d6e251883510b838926feaa9916ed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 14:04:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81d432a3f690bceb01a1096bd28c76d5d750a9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:05:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d735ddd327993eb438119a003739b023f5cba0df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 15:05:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df0aff46305c481d9fc1b47fb05179a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98afcbf198bf43449bf2b9817d3bf827 : SUCCESS in 4m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d14af280a8b442958a74c1c5e9aac749 : FAILURE in 38m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/72e27414e8d04caf8593c01e86c19624 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb8c6876f92c471dbac510f2e5bceb44 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65ee3b0064d84fc294667fc6d797c676 : SUCCESS in 19m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/657664dcc04640c1b2a46ed011373f21 : SUCCESS in 21m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/569855d5a1bd48b39fed870abba67915 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/780da98ddd8948cbac1d0f4be2fcc47d : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a243c7ae1f6462aa4ec13045deea4b1 : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47caa0e0b1dd465890a9431959244fd3 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/36da6f02d7d44905b8ba7d07ce639b94 : SUCCESS in 7m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dd5acfee8044479aeb55d0d3eb82ce3 : SUCCESS in 19m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9daab5dc9d5481ea357bf97a1663ea4 : SUCCESS in 53m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/994de37c518b4d239a39fb985d577b58 : SUCCESS in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6ec7d4ac4a4e77b6334993ea34a6b4 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca2006b37bbc4292ba18930bc83ef6f7 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ac49f464e4eb3e3e198c251b1709a61a5158da34","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-07 17:01:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0daccd032ef866d5de7d7526e351fa60e53fea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:34:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255e9d028e8f44baa1c76daf41d87719\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e5da5b738f4a43ab247fa076c11bfe : SUCCESS in 6m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71eec66f4d3148dd91b062b4cf8aeb30 : SUCCESS in 27m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7decd1ceb188499f8e0fadced9ff3926 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7bbf0808f80400a9ab10eff52600b18 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21ad46c7ce3747cbb3ba8589297856c9 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e80a9e2f03134ad99f88b69ff324fcf5 : SUCCESS in 3m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7102a4af9bf6436c92ad80a382ce9be6 : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f46845a4997541178c5a168f6e1fe528 : SUCCESS in 7m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0a84bd0218dc463380054d41e69a3d8b : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8fcdde3db8e84004be75d42960251df0 : SUCCESS in 12m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75fcebab65ea4724a5be8807cea0159c : SUCCESS in 12m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/590c690183194f5eafcd13e8b19d1fe7 : SUCCESS in 12m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f59545fcb92445e8d2267d154010326 : SUCCESS in 31m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd4cb43f248842749e3d39f16f5a2a64 : SUCCESS in 29m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae53d7f99f5944cfb48902bfb1b4046f : SUCCESS in 15m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9166661f824d7ca0ecef1453da9885 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/935ca984e72f431aa61514e296cf62bb : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e2afd142389efb0095afc2fad95711971c9e3884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:35:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b70e388297fc2e711cd282e336d2214094cd84c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b42288da67f462aa76ebae4b0c9c6e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/014633f1aeae47149c8b86f4a7b866f7 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c9c549cb5a0ad2ddbc9709f65485671d2de2f68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:33:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a1495d5cebf45fe9426392c94c0a8ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6cb1ab1e83481993d16ef7890e1065 : SUCCESS in 4m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/930106bec5db43db99bf3f6ccf2da985 : SUCCESS in 28m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5a85726311f940cd8779ae56e472a9c7 : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b4c2d1a61354de4818600c184306404 : SUCCESS in 7m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6c039ad05eed4832a02a2f38f6d654a5 : SUCCESS in 7m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a0201210d584a0490b5257ab7967509 : SUCCESS in 13m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3c37d9e0edb74a28a2385859f0a77b5a : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/059c72813e694b4e8aedd95e970f44b1 : SUCCESS in 9m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/65fc46f5a9944637996d8218186d37a1 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4e189e8f0da4d7b9c717ad9f79fa122 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43be0381b2094018ab4b6768b90d3392 : SUCCESS in 7m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/593f20f04a924d9c923c656fa54ea777 : SUCCESS in 52m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6d7fe5cb87945a1804641759dfede2a : SUCCESS in 25m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/952911dc6aae4b54bc74be101b3bc276 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a80bb2a4f01489095dd1163cd85e7c3 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80e64afbb9fa455d996b6ac0172394b7 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f87ab916e73132db63008b5aed0775bfe2fc9948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:33:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a0d6edf5512f7c19fcb7875ada8dcdc2a30d1911","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 18:35:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e41a7920c7463ba3294d2d6e521084\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05183ae3d1f044669b7e870e481de1bb : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/03004c64df6344d9a297971ef7397c5c : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cdd3915c4750aeccef956f5299870182eb70163","revisions":{"0cdd3915c4750aeccef956f5299870182eb70163":{"kind":"REWORK","_number":1,"created":"2025-03-07 13:05:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/83/943683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/943683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/943683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/943683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/943683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/943683/1"}}},"commit":{"parents":[{"commit":"59611253f9adadab95582251ae0c0fd8ad71fe52","subject":"Merge \"Remove notes about new-style classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59611253f9adadab95582251ae0c0fd8ad71fe52"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:49:51.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:52:21.000000000","tz":0},"subject":"Fix FakeSwiftCall partial env copy for None values","message":"Fix FakeSwiftCall partial env copy for None values\n\nNone values in the request environ should be copied, not replaced with\nthe dummy value.\n\nRelated-Change: I332ce724aa10287800cbec8ca21aacc3bbd3c22a\nChange-Id: Id78ac1cac833f9284946cd364b3f95341afcb235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cdd3915c4750aeccef956f5299870182eb70163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cdd3915c4750aeccef956f5299870182eb70163"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
