)]}'
{"id":"openstack%2Fswift~943684","triplet_id":"openstack%2Fswift~master~I73fac245c65c48d45e4919c6f15688d482148407","project":"openstack/swift","branch":"master","topic":"p-fake-dummy-value","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-11 15:53:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73fac245c65c48d45e4919c6f15688d482148407","subject":"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton","status":"MERGED","created":"2025-03-07 13:05:07.000000000","updated":"2025-03-11 15:54:32.000000000","submitted":"2025-03-11 15:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"943684-p-fake-dummy-value","meta_rev_id":"e22fadf7cbfe4009c7803572792130c90972748e","_number":943684,"virtual_id_number":943684,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-11 15:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-03-11 14:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-03-11 14:49:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 13:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 14:49:10.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"e9f8aa8362844afdc26ddaee270decf2288718a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 13:05:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc3cb8a733cd25c37acc99ca749835af28f8e3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6deae8a1b144b49b4af445535556ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da12fd98228c4b4680e6cc59ff8cad22 : SUCCESS in 5m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ceb5529df5247adb57f3afd4f58df05 : SUCCESS in 35m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15acf1c4e4b24a3c9e5a008d3f14122b : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d44601fcee134508adc224284b5ca261 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55c60b40d3d847a58cc89a741ffef13d : SUCCESS in 23m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d51f9e70574494b9bb5a82e8a4dd066 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00d55234c74442e0b72389287cc4c465 : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2633c281ab3446728a49b53b23b0c79a : SUCCESS in 22m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0d33299ab413408f92df551ebfbc9151 : SUCCESS in 24m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8074539a16f44e7b9ec5d5e100d4da39 : SUCCESS in 7m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9aaafeb921ac4ff7998b26c4d05f92d8 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb54bad5826742cf982703e8b2d1e572 : SUCCESS in 18m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/263e6d632dc94eaeb592abf7aea07134 : SUCCESS in 37m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fb1dc1aa39b48be81d623d2d491f202 : SUCCESS in 36m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/349977a0408e49f19a3aa67d9891e6ba : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10b7f17c17b45ce85b9cdfffbf6a41b : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba5849587d234c2c9652d8a3af1b3d87 : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b3ef09ae4556a87ee6e3abae0b68da2572b7ffa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:54:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dc5c9f582404c9b9266c9b58f6c23ed\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60c91ab22b974d3f9f3f09d2a4b2382e : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e0ebcceebd5307f249e88c8d7b6b29c8fb480660","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-11 14:49:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"17252036b9793796c1baebbc5551c475cdbb1c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 14:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c2ac6ab2b7091fba1aa60ffadae8f0d8e77578c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:53:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a8a706592b40e8885217c72deddd29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb979cf109034e9fb1987f6e005ff168 : SUCCESS in 8m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fbcc9144ac045589afddf5c33340c5d : SUCCESS in 58m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e5d8d7866feb433eaeb57137339e3968 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f7f5d154246d4111a7154065595075db : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/381f58b4e00b4e7bbd23ca7f676c8f05 : SUCCESS in 27m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f8e4078c94f412cadd85ca8ff4c3998 : SUCCESS in 31m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa1819bc1f05446b9828e39867c7ffd5 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd72ce447b6a4cdc80a4df3ecdee589b : SUCCESS in 7m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b67c77652ea144438509d4ac63030ee4 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86e76a0f7bf1434abc03fc31f9169bf6 : SUCCESS in 5m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0cd906f729046bfb61f18e33d7c4d4f : SUCCESS in 7m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b022da00a294ad9ad4a54c17a3d2136 : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d51d0f52f9d943529ec3b0cd2767f204 : SUCCESS in 46m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a904b1097bd745b0acb6c951371aa348 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef720ac496914744bebd86fd4a1069aa : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1811c3328f04362979938e6ce5532ba : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b06e295133d8ba6aab6077a71f5bce97dcbdbc77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:53:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e22fadf7cbfe4009c7803572792130c90972748e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:54:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fe79ad577c40848d067fef75970725\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6406e821afb45f981e7375b5ff19b55 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/496a3a4417df47bf867415c6c86e7094 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f74376e7e42cfdcd5cc19b4e2f3c455d1fa2966","revisions":{"2f74376e7e42cfdcd5cc19b4e2f3c455d1fa2966":{"kind":"REWORK","_number":1,"created":"2025-03-07 13:05:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/943684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/943684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/943684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/943684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/943684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/943684/1"}}},"commit":{"parents":[{"commit":"0cdd3915c4750aeccef956f5299870182eb70163","subject":"Fix FakeSwiftCall partial env copy for None values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cdd3915c4750aeccef956f5299870182eb70163"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:29:28.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 13:04:27.000000000","tz":0},"subject":"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton","message":"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\n\nIn test assertion failure messages the DUMMY_VALUE would be shown as\n\u003cobject object ...\u003e which is not immediately recognisable as the dummy\nvalue..\n\nThis change makes the DUMMY_VALUE show as\n\u003ctest.unit.common.middleware.helpers.FakeSwiftDummyValue object ...\u003e.\n\nChange-Id: I73fac245c65c48d45e4919c6f15688d482148407\nRelated-Change: I332ce724aa10287800cbec8ca21aacc3bbd3c22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f74376e7e42cfdcd5cc19b4e2f3c455d1fa2966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f74376e7e42cfdcd5cc19b4e2f3c455d1fa2966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
