)]}'
{"id":"openstack%2Fswift~943993","triplet_id":"openstack%2Fswift~master~I85ac768a701ceb26c96ec9074e37b2c4bcb62217","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-03-17 05:59:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-21 19:03:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I85ac768a701ceb26c96ec9074e37b2c4bcb62217","subject":"Add magic rules for swift rings","status":"ABANDONED","created":"2025-03-10 23:02:44.000000000","updated":"2025-03-21 19:03:50.000000000","total_comment_count":9,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"3ee0049563d3bfccc3c1748cbc86b262dbc4de03","_number":943993,"virtual_id_number":943993,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-11 17:15:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-10 23:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 06:07:24.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"695d29dfb1b61ee3e9f9ad022625a63b90dddaf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-10 23:02:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f30dea90e7f5732bff0ce78ec7f619d39a35ac91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 23:53:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699ba254dff241ccaf868c88fe07b1b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9af19a098b463aa2b8b27754291cf6 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/e0a44f55fc1a4d748a8cbd2070c2b095 : SUCCESS in 46m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6264ae104e9a49938d2c0de943e3fb16 : SUCCESS in 45m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76ff20dd90fd43e19e31c28ae004eb11 : SUCCESS in 39m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4ccf5ea037b489ead4536883b3d317d : SUCCESS in 49m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec1077e69f8d4af7bcefc70b274175f4 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c11a25cd33b4dbcbf553c333e4f54b6 : SUCCESS in 10m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af2fb7a8fb5d45b18bb54803fec50066 : SUCCESS in 10m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e072139cf82b405a9b20a592557c21d4 : SUCCESS in 3m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73b833d3ccec4689a254ce4799e22a8a : SUCCESS in 7m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/beedf9d4b6db425e83bb3bb5e9e6b1fb : SUCCESS in 7m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/99bff48deb1749be981bb98f5ac72a3f : SUCCESS in 10m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e9c0ed01b75445c2a48d473918d73676 : SUCCESS in 7m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/135c2a051f004b29876459087e6e0db1 : SUCCESS in 7m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c5e0da1366794b93b5ce8c4216bd7528 : SUCCESS in 11m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d684a90a9bd74991b712c21eb6524403 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22323cb4f4a74c7593ba2c9bf396e823 : SUCCESS in 8m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c305f1865c4f444d80fffb32bc0c89b0 : SUCCESS in 4m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/036edf458f6d4c40a9e977056f5f009d : SUCCESS in 35m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2124b83ce3314b04ad7ae5bf90d38e35 : SUCCESS in 24m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ca7bd864e774d3a99d0e16dce7463c1 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aec2cc321435424494c6ec42f468aaaf : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000bc3961ad44fe68e73210ff3775505 : SUCCESS in 2m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f590b338e0274149881aa2f618953830 : SUCCESS in 18m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86ae4cb0ca9442b4928221c37d02c32a : SUCCESS in 34m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c5afc8611f429986e455cfcaf58aa3 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0f695ec906ede01d9f7b0912dfb94761d89551f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 00:14:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3318dea405ee470295f7ec610dec9719\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf6ed4c821984c018fd0ee97b66e22e5 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"54c1717ea5fb19dbbe64856b01a0f390809e078c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-11 06:07:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"86f1348cb6b49e7d34f6236b628d9266dadeeb78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-11 16:00:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6788df513713561ec825f376165d4698f0ef5c7a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-11 16:01:03.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ff6a5203cf57bbbf5aedd7a66f9ba202604537f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 17:03:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac3a4f99ab7a42e6b9796ef829b980a4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f96b08457394ba9bc0db1ac7c82aa37 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":2},{"id":"383b4630671e6a612a72bf75864d97e970f05208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 17:15:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e01354ac43d41108c82425e50f00772\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008c3083effe4a3c9ee386b9a9733c16 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3939bf83aaa14a7290cd530e21e1c1fe : SUCCESS in 51m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cf7fa0f18254e4788268926369aaa75 : SUCCESS in 1h 06m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25a2d69428d84149b917dcac62e11b5d : SUCCESS in 57m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f029a154165b4d209ccb8ee39ab0becb : SUCCESS in 59m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/792cab2c8f7149889e287f856cfb1bd5 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82f10fb839104d7393633a16d607e46a : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87f696e344ab4a3aa3df38890ed778da : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4a581272db64880a74d1e38b0f4daee : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81f27733540f4d858111c85665b91f6a : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d269e17466e54d5abde6c493905e3a63 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c1aa65b6d3d49c39f39a2dbc097b6d3 : SUCCESS in 23m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e532a5157dcd464c9c43af7f8837dfd9 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f244c91a247482fa9d9254ba0bff17a : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aac5500145994716aa92706583bbee0b : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b00f73f94ea84a9e80b4983bce5f7747 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a346b0d4eef245f5ba0921c2c03603bb : SUCCESS in 13m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/762cec3c704343a3ba20b713808571fe : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a21e98fee09e4b5eab2897b0b3074e3c : SUCCESS in 52m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae26831752e045c0b88ba17fccfae7d8 : SUCCESS in 55m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a49e2632ec46434894b56542b3624426 : SUCCESS in 46m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a6675ecc1d046a0822efa0b85fa0605 : SUCCESS in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90854166240b46419ba16143726c14ae : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da3ef18471c64848b00cdb3eca9f2d34 : SUCCESS in 19m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f83c549ef9044a4ad64dd9df58dad41 : SUCCESS in 57m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da5edfa3ac0846c89b850ac8dac61a4b : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":2},{"id":"efbe5b9986dae2052152e6be5b31ae8952115319","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-17 05:59:24.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3ee0049563d3bfccc3c1748cbc86b262dbc4de03","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-21 19:03:50.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/834261","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4bec646f60c81445ca5c0061af1ee8237754eb1","revisions":{"2def45952ed9dc832fce6bd0e54373b033b0a28b":{"kind":"REWORK","_number":1,"created":"2025-03-10 23:02:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/943993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/943993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/943993/1"}}},"commit":{"parents":[{"commit":"0345b2b810c2ed35b4c17cbd9e4abb9102814b48","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0345b2b810c2ed35b4c17cbd9e4abb9102814b48"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-10 23:01:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-10 23:01:34.000000000","tz":-420},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nSwitch to uncompressed when we write our magic and version (to simplify\nwriting and maintaining our magic rules) before switching back to highest\ncompression. Add a magic file that can read that and give operators\nuseful output for new rings. All you have to do is append this to either\n/etc/magic or ~/.magic. Once done you can use file to determine your ring\nversion:\n\n  $ file object.ring.gz\n  object.ring.gz: gzip compressed (was \"object.ring\") Swift ring, version 2\n\nChange-Id: I85ac768a701ceb26c96ec9074e37b2c4bcb62217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2def45952ed9dc832fce6bd0e54373b033b0a28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2def45952ed9dc832fce6bd0e54373b033b0a28b"}]},"branch":"refs/heads/master"},"e4bec646f60c81445ca5c0061af1ee8237754eb1":{"kind":"REWORK","_number":2,"created":"2025-03-11 16:00:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/943993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/943993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/943993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/943993/2"}}},"commit":{"parents":[{"commit":"0345b2b810c2ed35b4c17cbd9e4abb9102814b48","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0345b2b810c2ed35b4c17cbd9e4abb9102814b48"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-10 23:01:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-11 16:00:11.000000000","tz":-420},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nSwitch to uncompressed when we write our magic and version (to simplify\nwriting and maintaining our magic rules) before switching back to highest\ncompression. Add a magic file that can read that and give operators\nuseful output for new rings. All you have to do is append this to either\n/etc/magic or ~/.magic. Once done you can use file to determine your ring\nversion:\n\n  $ file object.ring.gz\n  object.ring.gz: gzip compressed (was \"object.ring\") Swift ring, version 2\n\nChange-Id: I85ac768a701ceb26c96ec9074e37b2c4bcb62217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4bec646f60c81445ca5c0061af1ee8237754eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4bec646f60c81445ca5c0061af1ee8237754eb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
