)]}'
{"id":"openstack%2Fswift~944146","triplet_id":"openstack%2Fswift~master~I806d5602a0cf9be64e977182937e7f89d23083c5","project":"openstack/swift","branch":"master","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-17 19:41:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I806d5602a0cf9be64e977182937e7f89d23083c5","subject":"sq: move CRCHasher to checksum.py","status":"ABANDONED","created":"2025-03-12 18:43:53.000000000","updated":"2025-03-17 19:41:01.000000000","total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c3b0460208460d76c0ebb838dc85fae8540e8f83","_number":944146,"virtual_id_number":944146,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-13 19:38:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 20:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-12 20:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 04:32:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"e8be2ccbb7c725cb269a938feb529b2cfec9ba64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-12 18:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca88e1c3d309be47199e7a3f2ba5aa3e4e52f25d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 20:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/136c63ef56404e26a3a8a36a0067f276\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b330fda931cf4ae2b26f05994f88d474 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":1},{"id":"56220236793533ce9504692b6cadb6f213b3e73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 20:11:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a40bda63e12143bf9e2fee1b5a8bde5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1a20d7df8543fbb66e05a52d10d8a1 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/8de6bfa8fc4448e69f0a9d56b54c38cb : SUCCESS in 1h 05m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e369b25e470b4e929dffa655c0af2eef : POST_FAILURE in 1h 04m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d94207a197b4280bff8eecd94d57fd3 : SUCCESS in 59m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1756a30574ea4bf391f6174efc01c0c3 : SUCCESS in 1h 09m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b91e390531a458281037c2d230306e4 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73a9d54899b2458ea7782b6859d4af1d : SUCCESS in 24m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecedf69b6175478b8ca0e44a8894d603 : POST_FAILURE in 23m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b90925c3c5a54362939bb3faa875111f : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f9fe21ffbd34e5d97bdcebc7f882d77 : SUCCESS in 9m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61c42afeb4904ac9a8362b964180088b : SUCCESS in 7m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/603f03eb9beb440d8cb7654d8b7a5ca2 : SUCCESS in 30m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/832c89bdc1204371925a4c6d4ecc73e3 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90f047bad3ee490aa78bef2f2cbb10bb : SUCCESS in 9m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3e51d8fd45443f3bc077a8cd513c8a8 : SUCCESS in 7m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6104c01bcc4446e6abbacc78b9563fc4 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15017cea71394e9490403f4ec9ac87fd : POST_FAILURE in 16m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/354131b1a21949daade2e4c385fac51f : POST_FAILURE in 5m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fe17f04c2e04eea82785c510e98dafa : SUCCESS in 33m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30cc6773735744469eb3e21d8c08d25b : SUCCESS in 1h 05m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/163705c7560a4e8793f02c96012a3c58 : POST_FAILURE in 59m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e37c0c750e64b62ae7d42b1a1e746e5 : POST_FAILURE in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deabbc017e984829a26a0c84d9d468ce : POST_FAILURE in 3m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8334b57fa8c442fe98596314c11bf039 : POST_FAILURE in 12m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83f5361d10d647a1aca4962aad7f042d : POST_FAILURE in 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf73db8ce23e4001913087038f541ae1 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"032e1351b08a330cde99b9046529200334e5b8bc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-13 04:32:35.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f7c07c8b85604a86e600b86c2c896ea566de2edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-13 17:58:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"131981f303f4a0a5556b19644e03f6bc7b2f73c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 18:34:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5912e6c8f7b34f8e9b4ac1bd22433886\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e119b191ff6442089fbe65b38c31863d : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ec328a6c3e2db608de1f1a5d0e51c8b982fd4d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 19:38:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a30a91bdb05744f28b41c679039072b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e0454f91fdc4e2baf83ff7ce8356918 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a5d39f98b4dc438f929263f2eb500beb : SUCCESS in 59m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7eb1950de464a57b496b1848aa00f2c : SUCCESS in 28m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a47f6086b0b4f3ba95452eddd156129 : SUCCESS in 30m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/796fa992655049f49ff899ca98b6064e : SUCCESS in 33m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f58dd164d72d4147b5ab3d3960000e91 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc5c71bf243f49fd924e9106abf1a579 : SUCCESS in 10m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d297ac28c54e4c03a8fd350d9106dffb : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe237fc774da4772ad1aaf3dada70ecf : SUCCESS in 3m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/168a2980f08e4485b85eb66d1e53cac6 : SUCCESS in 20m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e938557b8c964b96a4071d95be62cea4 : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/282dddee0b4040e7ae4260ffdc49b7d1 : SUCCESS in 27m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/534d3ebb2076495d91f95f4c6892e74e : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba861e14bc9e4fba85b860257d19faeb : SUCCESS in 9m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/790970ecb61044f18dfd731433b3fa5c : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d59bd6545d84be2a183784f898eefb9 : SUCCESS in 5m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/676a9c0e0eb24fd38b6cf8caadb1e90b : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/15c7255b4c414ac895596b39e395e8ea : FAILURE in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/65bc72fab0624e59a56d6ca447a33e34 : SUCCESS in 37m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a354a8c9328f4dcd9493dd61c755f474 : SUCCESS in 1h 00m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6725ad8fa5c4e3781ddf1cb0996d95f : SUCCESS in 55m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11bbf10642954e9d96afd409aacc9107 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6843e1d86e4e2f9d0c7ec57ac2663d : SUCCESS in 10m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36970dd3e0594522a47110d21906b77f : SUCCESS in 23m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d203e3cc80404c7cb3d7b3108750c4f0 : SUCCESS in 54m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d427130572744542a6314fc2f88a21db : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cef52655af133a50d3f07ffcb8d7685a9fd50b0c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-14 11:50:44.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":2},{"id":"c3b0460208460d76c0ebb838dc85fae8540e8f83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-17 19:41:01.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5a18d1bafaa78b2ccdd1b4b132dac89e37b7173","revisions":{"75c62240ead0e070046b6fefdee2add2d74233b7":{"kind":"REWORK","_number":1,"created":"2025-03-12 18:43:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/46/944146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/944146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/944146/1"}}},"commit":{"parents":[{"commit":"d0a805570aed2afdf69590f1f73e8dcdb4aac3d9","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0a805570aed2afdf69590f1f73e8dcdb4aac3d9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-12 15:14:04.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-12 16:31:38.000000000","tz":0},"subject":"sq: move CRCHasher to checksum.py","message":"sq: move CRCHasher to checksum.py\n\nPut all the checksum related utilities in the same module and make\nthem more like hashlib hashers.\n\nProvide module functions to construct the various hashers, similar to\nthe hashlib interface.\n\nSupport passing initial *data* to the hashers similar to a hashlib hasher.\n\nAdd name and digest_size attributes to each hasher.\n\nFix CRCHasher.copy() to copy the width.\n\nEnable explciit testing of the crc implementation preference order.\n\nChange-Id: I806d5602a0cf9be64e977182937e7f89d23083c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c62240ead0e070046b6fefdee2add2d74233b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c62240ead0e070046b6fefdee2add2d74233b7"}]},"branch":"refs/heads/master"},"a5a18d1bafaa78b2ccdd1b4b132dac89e37b7173":{"kind":"REWORK","_number":2,"created":"2025-03-13 17:58:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/46/944146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/944146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/944146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/944146/2"}}},"commit":{"parents":[{"commit":"d0a805570aed2afdf69590f1f73e8dcdb4aac3d9","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0a805570aed2afdf69590f1f73e8dcdb4aac3d9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-12 15:14:04.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-13 17:58:04.000000000","tz":0},"subject":"sq: move CRCHasher to checksum.py","message":"sq: move CRCHasher to checksum.py\n\nPut all the checksum related utilities in the same module and make\nthem more like hashlib hashers.\n\nProvide module functions to construct the various hashers, similar to\nthe hashlib interface.\n\nSupport passing initial *data* to the hashers similar to a hashlib hasher.\n\nAdd name and digest_size attributes to each hasher.\n\nFix CRCHasher.copy() to copy the width.\n\nEnable explciit testing of the crc implementation preference order.\n\nChange-Id: I806d5602a0cf9be64e977182937e7f89d23083c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5a18d1bafaa78b2ccdd1b4b132dac89e37b7173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5a18d1bafaa78b2ccdd1b4b132dac89e37b7173"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
