)]}'
{"id":"openstack%2Fswift~944388","triplet_id":"openstack%2Fswift~stable%2F2025.1~If3c1f6b6ca594519c025b054fafdae51ab32b00e","project":"openstack/swift","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"If3c1f6b6ca594519c025b054fafdae51ab32b00e","subject":"Update .gitreview for stable/2025.1","status":"MERGED","created":"2025-03-14 14:26:45.000000000","updated":"2025-03-14 17:05:50.000000000","submitted":"2025-03-14 17:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944388-create-2025.1","meta_rev_id":"ad63380455a88d9257d91bad8a1272c227071945","_number":944388,"virtual_id_number":944388,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-14 17:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-14 15:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-14 15:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 15:03:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-03-14 15:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 15:13:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b7d8b451cc2cea58c63f7fe4657f0e6ba37d662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-14 14:26:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0147fcbb791e2786ddaa5dffbe87ca1bda730108","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-14 15:03:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a212862edc18290706cf75697b06be79f8df2626","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:07:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49486063be7345f3aec966ebd9a9d43c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/319c404ecd7640e0a5b024d1a2f7f9db : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2cf0d04774727f929fabbee145b0fc8bbaed3861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:13:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e339d94ff04cdb928169700505c08d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b71f76f5fb4406d8af7f64178dcfebf : SUCCESS in 4m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcdd8b91d85541fba3f60c27868252ef : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fac2bf2763d14f2a8d5e3970efcd47af : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd8113b293b148e78648b26d89af07ee : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7cf584770fa247a3a21047ef827f0a17 : SUCCESS in 29m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74d53270e97407380591b2f9d51c6d5 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e933fda84e940d0b21b9bb94a0ac3c3 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bea7e7d20a0d28a72db5f53009df07d0ae5c98cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:13:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d55745ce475426a5e57b4527a201b7d4f5942c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a2cee862254a27b1be6ba2cceace6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e03710d8096476eac2427071789f098 : SUCCESS in 10m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ef2140ef7f0d4a8aa9dc6e51d360a81e : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc97bc7ea9c4ee6b1ebd73d3daee052 : SUCCESS in 7m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbda7693d93447585da0f9142002726 : SUCCESS in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3297604952049338e9ab317220902f2 : SUCCESS in 38m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da52f537ed914dfea821ba89d4fe0bfa : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5095bbfa4ec04cf1a18f78bb5d072d8c : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/528ecc80ba15413a88c657c7b96188c2 : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38fae1c18cb44a8e8efe3eee5968b98b : SUCCESS in 7m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/888098af098e4d8aa5c730ca9e5cffb7 : SUCCESS in 6m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b45c0c1f04fe42a69335807d6ae52774 : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe38ecc7b1b4e87b71e3d2861896464 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"017729a8d695698be9bbfa18e79f0dff097c9d73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:02:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad63380455a88d9257d91bad8a1272c227071945","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:05:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92964c91bb0544e89f2350f3c679b9ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f23fc9361249f79a4c5484d328354b : SUCCESS in 2m 09s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/339da3b2aadd47b5985ee8166bc770a8 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31d6e6a83a7587203c5b4058578dda52482cfe27","revisions":{"31d6e6a83a7587203c5b4058578dda52482cfe27":{"kind":"REWORK","_number":1,"created":"2025-03-14 14:26:45.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/88/944388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/944388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/944388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/944388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/944388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/944388/1"}}},"commit":{"parents":[{"commit":"d6ac3a261fec6420ff2f15d59921d5e03aec37cb","subject":"AUTHORS/CHANGELOG for 2.35.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6ac3a261fec6420ff2f15d59921d5e03aec37cb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:43.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:43.000000000","tz":0},"subject":"Update .gitreview for stable/2025.1","message":"Update .gitreview for stable/2025.1\n\nChange-Id: If3c1f6b6ca594519c025b054fafdae51ab32b00e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31d6e6a83a7587203c5b4058578dda52482cfe27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31d6e6a83a7587203c5b4058578dda52482cfe27"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
