)]}'
{"id":"openstack%2Fswift~944389","triplet_id":"openstack%2Fswift~stable%2F2025.1~Iea166f7ab0c027542ed89c0ffc793e3b31029043","project":"openstack/swift","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"Iea166f7ab0c027542ed89c0ffc793e3b31029043","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","status":"MERGED","created":"2025-03-14 14:26:48.000000000","updated":"2025-03-14 17:07:13.000000000","submitted":"2025-03-14 17:02:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944389-create-2025.1","meta_rev_id":"dde644c332cd7ebea8c9250ef135be31637da467","_number":944389,"virtual_id_number":944389,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-14 17:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-14 15:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-14 15:03:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 15:03:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-03-14 15:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 15:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8e88245f80f846740a0a4728d2f4e4ad507bc9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-14 14:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62dc05415bff8bb22c2f6eee7ca1c705cff5f619","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-14 15:03:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cb1494ac17f4af416d21b3887d40dfcd89e28ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:05:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f81e43835d44e059e2c3d8f63eba18e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dccdfd0edae4be289457e3dbd8637dd : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":1},{"id":"eace55b85406c4bcf5eaa743245324a5687c59ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:16:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ffc94ebfa1495596df168c0714b6f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355fc66d047b474280cb73b3c207c7d4 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa8bbf7dfe2f4a299208fca4d8f7f401 : SUCCESS in 6m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0df1058da8544b69a56285ba5677652a : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ac834f404cf456f8b8a1e9304c12e28 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b29b6b1a4235464e93b1763df796feda : SUCCESS in 22m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d21475297bfd4b8c8328823f6292d09f : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/247e7c18a21c4f0c9d5269ad1bc0aedc : SUCCESS in 12m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c79df6a5a0b4f97a4ae00544c649f66 : SUCCESS in 7m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f2a61e5c6d224a7e8dd6139388c7ca24 : SUCCESS in 28m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5349e8a435024f9b997aa799ad0dacc8 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0582d43004064bde9b006f214eb2a503 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/de42f384f85c42e4bdb5e4617f745f24 : SUCCESS in 7m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fe5c26d69a24af3bfde93599927b6e7 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47b896bfca3e402aa995e3d1182fbfbd : SUCCESS in 7m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d61c6ea8c0604c449db7156c33fb9c9f : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e05d6435edc84c32971ba8c235ef3a33 : SUCCESS in 32m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b5a5a900ca14dff8fe186a08cceaf81 : SUCCESS in 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d787719f0f45b9be218b26295029b2 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eedefff3f35a44b88bc67acaf088d82e : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d41458ab755563af8983f7a647fe7e4985209374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:16:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23a9628f6b8f92cc18694e3d4c138ad614d220f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:02:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb03af695c7a4d86841169efa2c9b356\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849ea53b5de945938b3112a6454210f7 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6125f533c8a74f4cbb0f960cfe169b2e : SUCCESS in 7m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/958722b17b4d4ca298fa7d8856b11f53 : POST_FAILURE in 8m 36s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b53856d6e6604d5ba747c1e4b517b0cf : SUCCESS in 17m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d55eeed2532490b9ebae768a888a785 : SUCCESS in 20m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7201d2afeb20435a9ea6eb9ee3702f8a : SUCCESS in 22m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b4aac4b5c9f7428bbe46b194cfc87a9a : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b383d3aa4a6f4c509b4d1dc16998b735 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4b806ae96c9a447a8a93b256bbe4592c : SUCCESS in 12m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/04d3b8255aa941f9b35669599de49f40 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95e98c42ab0f42068a0ff03f6de26589 : SUCCESS in 3m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b40e0cc724b45938391976625bd4758 : SUCCESS in 53m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b065c1dbb954e71bf4fbba4e8d1aa8d : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a06897c6c644bebe2bb7730e310840 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4fdf7cb28ec4f04bd08cd3e518ef4f7 : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":1},{"id":"49247d5df34bc34220c6ea05645be948ff257617","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:02:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dde644c332cd7ebea8c9250ef135be31637da467","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 17:07:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93e59be9d5343b7b168a6bf150017ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854c73d112de46a783333a90054f3ce5 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1101642c937f4ba98e9aa5b9228bbf8e : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c507224b53c24d37ac25c2b5485c48aa : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"095779bf41ffb090747aee5c8670328a8cdc08c2","revisions":{"095779bf41ffb090747aee5c8670328a8cdc08c2":{"kind":"REWORK","_number":1,"created":"2025-03-14 14:26:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/89/944389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/944389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/944389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/944389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/944389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/944389/1"}}},"commit":{"parents":[{"commit":"31d6e6a83a7587203c5b4058578dda52482cfe27","subject":"Update .gitreview for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31d6e6a83a7587203c5b4058578dda52482cfe27"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:46.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:46.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Iea166f7ab0c027542ed89c0ffc793e3b31029043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/095779bf41ffb090747aee5c8670328a8cdc08c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/095779bf41ffb090747aee5c8670328a8cdc08c2"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
