)]}'
{"id":"openstack%2Fswift~944449","triplet_id":"openstack%2Fswift~master~Id8c40506497ca166196fbe7b32d05eb2742593cb","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-17 17:15:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id8c40506497ca166196fbe7b32d05eb2742593cb","subject":"sq: catch InvalidRequest and read single byte in one place","status":"ABANDONED","created":"2025-03-14 17:59:11.000000000","updated":"2025-03-17 17:15:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c564916de8a0df22bcf8d5fe2cea0fc79b0bbd9c","_number":944449,"virtual_id_number":944449,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-14 19:38:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 18:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 19:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"256be7bfa57747d234696ff743ce2ff11bb17081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-14 17:59:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f171ab97bf6325137d4de34a138ba7d2b120800","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 18:54:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19fb76ba34b64beb8678c35f47932df8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a20d8ff4aae64b3495c75b3c5bb7880c : FAILURE in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6fb9c9b4e57a4a0b67982f4ff6a7d670f1d41638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 19:38:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d12768c89f24dd59fd6f993762ac2f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbd5bb926c6464eb9546e0b6b6ba350 : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/fd4fa2393e92491aa384c0e40df4c449 : SUCCESS in 1h 11m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3e89c6922614a7f8a4fe9789fb1e6d0 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59e18f0df29649b5a000efb191bf40c4 : SUCCESS in 1h 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6210b2ee44740eda163a0222878f4c4 : SUCCESS in 1h 15m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7308de8c0c748de870fb64b3fbac887 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69e1ffdd73c44193bcc40512801ae8da : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d536f9f77004c7db0d7881862ccb27f : SUCCESS in 28m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/166864f2029f43b6ac78926b79befe4b : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0df2cc18c9ef4c329c92352e99405c38 : FAILURE in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11149b831f814c958315a08bf78d40e4 : FAILURE in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/56bab28b3ad943728605e63b64943354 : FAILURE in 27m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e5bb1afbae984b8695bb44880e35ba3d : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/51ff15e05f62438d941db3aa834cb02e : SUCCESS in 13m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ce585ff7dc7f41459f1a01247ff9a896 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f813fa8844f045fc9cb4286ff40ffeb2 : SUCCESS in 5m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7829652cba14095b6be4c05a2a33e82 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/453439ad889045c9b5069df660453a1b : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df6bb9081c4a4f19aa0cf0644077be0c : SUCCESS in 49m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/33679bc98a91407a86561900b0dc5356 : SUCCESS in 28m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/486ef2974fff4330a895b772cf3799c6 : SUCCESS in 19m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdbb7d0bc2964e05a57072c5f9cc68a9 : FAILURE in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df493bad582b442f8e7439169bc93e2c : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d27f4bf1c544669827789fd211924b0 : SUCCESS in 9m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0dee4af068040a0a798751040e1e36a : SUCCESS in 40m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/495522e8f5884262ac9dd3d06c320c65 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c564916de8a0df22bcf8d5fe2cea0fc79b0bbd9c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-17 17:15:28.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/944073","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b7ccd63170b568a202c392aefec23e684952b9d","revisions":{"1b7ccd63170b568a202c392aefec23e684952b9d":{"kind":"REWORK","_number":1,"created":"2025-03-14 17:59:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/49/944449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/944449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/944449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/944449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/944449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/944449/1"}}},"commit":{"parents":[{"commit":"87e1ad0fe0805684981dc11fa8b9b8e2b2f64922","subject":"sq: raise an S3InputError when trailer crc is bad format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e1ad0fe0805684981dc11fa8b9b8e2b2f64922"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-14 17:42:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-14 17:49:45.000000000","tz":0},"subject":"sq: catch InvalidRequest and read single byte in one place","message":"sq: catch InvalidRequest and read single byte in one place\n\nSome of the tests here should be useful even if we find a different\nway to shut down the connection.\n\nAnd if we find we don\u0027t need to do this at all, then there\u0027s now only\none place to rip out the read().\n\nChange-Id: Id8c40506497ca166196fbe7b32d05eb2742593cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7ccd63170b568a202c392aefec23e684952b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7ccd63170b568a202c392aefec23e684952b9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
