)]}'
{"id":"openstack%2Fswift~944450","triplet_id":"openstack%2Fswift~master~Idd854e2aca72f447cccf14504e731fad3d1c1ab2","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-17 17:15:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idd854e2aca72f447cccf14504e731fad3d1c1ab2","subject":"sq: move digest validation out of ChecksummingInput","status":"ABANDONED","created":"2025-03-14 17:59:11.000000000","updated":"2025-03-17 17:15:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"07c1624962ca5691354861a39cf5e1bc279f345a","_number":944450,"virtual_id_number":944450,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-14 19:45:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 18:55:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 19:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a47f63d00b938889a34853a193d8edec85a7f503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-14 17:59:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f15736214717847b34b3264683c7cca6ccadd11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 18:55:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd7e7f18a814ef3ad8ec6e3527620c1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/245bc95dfe8b44ba9cd8dad10374f406 : FAILURE in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6a9fc709cf19b237ab97d8c7dd0755a4ab21af24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 19:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8736244a1bc5477c84cd52b8cee0a280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca7601b8ce54f09ac6b6041fdcfc79a : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/6e7391cd825246a58e5f825b8f2e884a : SUCCESS in 1h 07m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f6322bee5184cd798756f48271421a7 : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/156cd588d7a840e9997de738cf8a5422 : SUCCESS in 38m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2af2c917245e434295c8bc78178e8197 : SUCCESS in 52m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbb500b63fb44f6ead1714fd91a25af0 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca2b0e8b1b8940b48998edef8992e425 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/534f53d713a343d883c6387486f145ed : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ff8ed8fa71a46a1a3c4d754621d2c0d : SUCCESS in 12m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/306b0cba97c640d2ba6e1ec9be724cf8 : FAILURE in 20m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/976a7ae8c8534f2eb9937bfdd8e7b349 : FAILURE in 7m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/63bec68f0c4343059241ba548f7ba590 : FAILURE in 13m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5aaec34ac09a4e769471da4937d3f17c : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ecd8b7cc36604aa3b5d3d369b7b6c98a : SUCCESS in 7m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e3cabf8513d04b74814c1106fa820468 : SUCCESS in 14m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/04eac3b6538b47118996e3cdc2cfd51b : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bea288027d17446d8fe93ab422457465 : SUCCESS in 9m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/00ab56d53e7f469395c2e3d7db7c0284 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ff47012b5b04dfb87bad3b9deaf531f : SUCCESS in 50m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4430d1c9aa94f94ba344dff5d088015 : SUCCESS in 34m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6bd96d097be54165a63d7103caeb89ef : SUCCESS in 46m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c5353561ab24f9bb17f6a1eaebb2dd0 : FAILURE in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f234083328404d8e8c3975e8fa8425 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02f19ddb55b440babdc365a2659dffd5 : SUCCESS in 20m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8220416dbda44ebc8632875b05ccaa4d : SUCCESS in 47m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca1b6cf89a9c41faafb4b6b9c2e4f2db : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"07c1624962ca5691354861a39cf5e1bc279f345a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-17 17:15:22.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/944073","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddac040ad34a979df3c37423b40102d94a457861","revisions":{"ddac040ad34a979df3c37423b40102d94a457861":{"kind":"REWORK","_number":1,"created":"2025-03-14 17:59:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/944450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/944450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/944450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/944450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/944450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/944450/1"}}},"commit":{"parents":[{"commit":"1b7ccd63170b568a202c392aefec23e684952b9d","subject":"sq: catch InvalidRequest and read single byte in one place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b7ccd63170b568a202c392aefec23e684952b9d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-14 13:22:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-14 17:51:01.000000000","tz":0},"subject":"sq: move digest validation out of ChecksummingInput","message":"sq: move digest validation out of ChecksummingInput\n\nChange-Id: Idd854e2aca72f447cccf14504e731fad3d1c1ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddac040ad34a979df3c37423b40102d94a457861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddac040ad34a979df3c37423b40102d94a457861"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
