)]}'
{"id":"openstack%2Fswift~944506","triplet_id":"openstack%2Fswift~master~I155572bc5f4d955c918c951d6d2af9dbbae0136d","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_object","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-17 15:38:38.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-05-19 12:20:01.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-05-19 12:20:01.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-05-19 12:20:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I155572bc5f4d955c918c951d6d2af9dbbae0136d","subject":"Object-server: add labeled timing metrics for object REPLICATE request","status":"MERGED","created":"2025-03-14 19:59:35.000000000","updated":"2025-05-19 12:22:03.000000000","submitted":"2025-05-19 12:20:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":109,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944506-tmp_labeled-metrics_object","meta_rev_id":"355117ac981dd39cb9a8dd01075109be19b6f19a","_number":944506,"virtual_id_number":944506,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 12:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-05-19 11:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-05-19 11:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 20:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 21:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-18 14:26:26.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2025-03-18 14:31:55.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-03-19 17:38:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"eaeb8cca3eb6af2f73fae6704c88bf86b1603752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-14 19:59:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5756bd6eea470388f3fe82c64ac42d6f451dbd69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 20:43:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7739cfec7f64293a65ba23fcebcd749\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c523a9713c454f68ac5e45ca4a8b5ce4 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8d8cc176aa0c91a822cef2b3ee4757496b11c1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 21:06:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07cf8fe9b4df4a2ca9fcbcb5a9929330\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546405a1078b40d3bada71675584996f : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/fd48caafd96f44c4a6daa330687aca44 : SUCCESS in 43m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23c46c0be7494d049bcf97d02c00059c : SUCCESS in 33m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8109fd1a3d624b0fb7a4e572ae918f24 : SUCCESS in 57m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/460334ceccef4b348d34dbcb021f4651 : SUCCESS in 48m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cb767dd04ef470b8613b889f22c03d6 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a65316d68f9491fadd4d759fccaf283 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3a66aa0742d447a95b25c75cfbc5a9b : SUCCESS in 16m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/244be23fe7a64b94896949683bff05c4 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/567360bf1a264334b28c90d6b6cda1ce : SUCCESS in 7m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e88bb0a377dd40b0bfcc457201e1dcdb : SUCCESS in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db1cf7c3c759461abb662365391fa6a8 : SUCCESS in 30m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94eb1396b3ef439a847e2ab353a7f7eb : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14b5f08e2de14f299a79f4d1d64473bd : SUCCESS in 12m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1189666d21f4209a7188051e3dfbab8 : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9013fcaa04f041d8ba4e09e97d054070 : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fec412dfe7f4525842474193337c8f1 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0aa02c66ab9643a4a3ebaf0db9afd0bf : SUCCESS in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf80f416428d42498b325a79b8ab2bbb : SUCCESS in 36m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/891e91285c3d4b468fc0b185ecfe286a : SUCCESS in 32m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c9033a9272a45a295b8de358edb8919 : SUCCESS in 51m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44edc7bc30b7499fa68ee7f75db4d5ae : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c69058b4764becb75bdc7adce7d2b9 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d989b1f430e4aa197bc102c24da15c5 : SUCCESS in 18m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/171d21b337b7477787bf1f5f391ced2e : SUCCESS in 1h 00m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6206d716b2754d179ef72afdef7ac663 : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5fdb79c43ee73a974baa2e97299aeab64e0423e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-18 13:58:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9c6997a8157df29097218fdd3b27b68f3c3c1f2b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-03-18 14:26:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9b03070f601d4661592dc3459692eeeb21cccdb3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-18 14:31:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d35da674d1d53f145a0f9d072809396345cabf50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:38:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2df7b172a544be3a14ea33bc5dcffd9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/930742f6565443ea924da7fabc18b9fd : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":2},{"id":"41784a9e028fdbfdc922b15329bcb1f21a897c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 15:18:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86372e221ecb474486bab7fb04cc0fcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6565356ac8485e888ff18764ceef74 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/556c9b3f507346c286b54b0639b3e5e1 : SUCCESS in 1h 05m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20c0a29f9eda4898b335925b3b9e5109 : SUCCESS in 54m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/713374070d594515bfab6a36c9070957 : SUCCESS in 53m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1254b13b60454e24986356f0ecb98e2e : SUCCESS in 1h 01m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/622ccadf9919438ca2d443b881cf472d : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2effb0e4254946a487b61c6cfc592859 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e400e062532d4c3fa493c98c003525a2 : SUCCESS in 12m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69f69496e20b48d3b45c2dad84ff5a01 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1367f0fa4edb4b148f7168ecca501bbc : SUCCESS in 16m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e94e138ccb264f8d8b1d34b00ac646a5 : SUCCESS in 13m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/89a91ac5fedd4525a2b785e245b232ad : SUCCESS in 29m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd54aaf38f654c32bac7d5e65e742a55 : SUCCESS in 7m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c94c3484e1b433f9344e59b207f264f : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aaa549c35ba849d3afdea2b1bf06fa45 : SUCCESS in 26m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb88e3112e72483cae9fe1779d4be7ab : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca88e82e7e1a4a7284a4456383d80526 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe956fd4fe654e478f8c92a7164ef596 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/baa97f903dd54bd5a22894c6a76f7bab : SUCCESS in 1h 06m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ddd5a6562f74872aee9085e9268e6ba : SUCCESS in 1h 12m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b241a796ea544089b324bf6db1a75252 : SUCCESS in 46m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c712767ae6f4153833eec47486f75e6 : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deaba7cc963b473b8d52ffec3f0b852d : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb82ea9f4f144c6db3e319ae11e1182e : SUCCESS in 10m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7a81bea305546a68b8e8091554da98e : SUCCESS in 58m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66fbc02da2f448f78e1e0a81b6098fc0 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":2},{"id":"679b60bc3e715eb727a2bdef070ee3f24c378e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-19 13:59:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e83f920d50f176a2353aff2f9f95db6f0c5e5780","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 15:04:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39600671b7c04ee5bffe9032ef0bfca4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ec1883767a847e5ad5554741ad4d9c3 : FAILURE in 55m 46s","accounts_in_message":[],"_revision_number":3},{"id":"c403649764e4ce411bb59970992255aea8d55282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 15:11:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ba7a3488a6d4d1c8d2c2a7a55379e0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966d9b47a9f4457f82dacf7515e76c65 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0a764d9bd57d4b9abd08c7d5163ed8bc : FAILURE in 37m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3450069cb9ff44a4ae4d11c9224ae4c9 : SUCCESS in 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/067b9b1eea6a4018a37e0c88e58e2375 : SUCCESS in 57m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2db5322bfa74c0fa2d06ff2593b7a58 : SUCCESS in 35m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/265b66329a2b4659ab3ba229942805c1 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38a698397f504c428129aea3f6759a0f : SUCCESS in 23m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbaee46e9cc941d1b54e67b338f2d3c2 : SUCCESS in 17m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de4b9b90ef6041b18d46fd3664c9032a : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b3029106a2634274b98c30dfaff77365 : FAILURE in 25m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25c99070b50d48c0af5cbd656cbfe373 : POST_FAILURE in 22m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c1e89e098e7544dd88222767bd783ce9 : FAILURE in 33m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dbfae8c2558f46eabb0e98cd100efd79 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/031615564b0a475ca3ddba92ed15f83b : SUCCESS in 8m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b929c980d9d4ef197c0450dcce5ccd0 : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c26b41de423d42b68e0c6421f90f1d9e : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fdfb9d12c014c1ba36dd6de13f13af1 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/176bab2627ee4cd2b1679657b90ea075 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c13e63ad32704a7a90b7b57efd236a5e : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/666a02e771df47f083531458f73a0b4c : SUCCESS in 50m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87d1849828954ec0ad926815d5d93139 : SUCCESS in 44m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50ef7230e36540fd82c54e0757e1977c : FAILURE in 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364d3ce9f1e24c6e971fb804eddc9356 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/037186c5f44a4373bb916ad0f032fc32 : SUCCESS in 21m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a27d2947a4a54342a4b7f711a02a8b7e : SUCCESS in 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/142cd157cbbb4ecfbb2237d3d38b94a6 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a80e6625998b33ad18c2d3c3eb9eb21f33fba247","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-19 15:26:36.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"21f772c73bea03df20fad000411fd164946353ef","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-19 16:35:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b570b3f00408162e1dfe472c9ffd408fcfa8ba98","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-19 17:38:02.000000000","message":"Patch Set 3: Code-Review+2\n\n(18 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9aa1e256315a0455846023985cb45bc17b87b677","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-19 18:45:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c0c16b6d1001ef6e779990202d74c929024a9702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-20 13:58:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c97480a30af92aa19f1d09cf69b0343465d461ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 15:11:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dda65b58d1549dfb7d2a88c7486b0fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed742a4543d497482bbbfc682519be3 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/17e569d098354e3591dac0192be2a7e9 : SUCCESS in 50m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b123a43477b4656977c7443dacee05c : SUCCESS in 46m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54a512688e97453480d01bb88982b7d3 : SUCCESS in 51m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9bec3d68cbd4a52bb178e9f1ffcf99c : SUCCESS in 48m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9484f98c7a7842c7ab431992abf980b5 : SUCCESS in 25m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef0b97b839f64ba892a444ef76cf0893 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f744a5d16d9049fa84d9c10d5ca63bc0 : SUCCESS in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7588fb9787e343c386c55ef4a9f64329 : SUCCESS in 11m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6198aeb4c1bb43ae813779768aa3e68b : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7eb0fab0094d46af83bc964b3d0f0a : SUCCESS in 17m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3f478d9631a3435592c1a7d8f0398a85 : SUCCESS in 33m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/03bf02a2c28e49cdb935f96db5eec38d : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/658bc39d53dc4fa2835d0f558bb85fea : SUCCESS in 20m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1e2419f637104323bd2b61145d7c2501 : SUCCESS in 18m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/912c50f8e9404d658b5345d0c8246e3f : SUCCESS in 17m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d10acae3e27e4efcb2737c7ecfd7872a : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fbd42973d78424ca615b012b9f6c893 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce44feeb052d4fdaa898c71db43e028b : SUCCESS in 51m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab32c2bc4e974815a4045b0b4d7a3b07 : SUCCESS in 43m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b8d2fbb10714f278dfca11877481984 : SUCCESS in 49m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5787f41a99042559123a87560c1a7fd : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1248d822b1d4af3818875c772f94652 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7dc8a9f7c92743a48e2a00d466ac8300 : SUCCESS in 22m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54732f6d825d48d2b6c6aa96bbe502db : SUCCESS in 1h 06m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0cb7830732a4600b0019054f158253f : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":4},{"id":"4cd465eb6ab1357b0665f5da82bd06b30ec1d72b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 15:15:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcff3fed817498cb695d0a4e3209314\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e1f5b9eaa754cc9975eff6047df58dc : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":4},{"id":"14d23696909b4126906355b0735654a990f8b1c2","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-20 15:32:43.000000000","message":"Patch Set 4:\n\n(14 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9f7808f56a267408e6add08cbbe718e57e17f571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-21 13:59:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7cf79b280180abe3ea7189b8fe481498c729f12e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 14:48:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d087388abd5141fc852eef7cf6e2792b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d30ef045314d48f7897a8b01dd1c49d4 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":5},{"id":"b475c740559d1643b7e372beda75191147421256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:00:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338e970e354541be9617567fc9a62938\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7445f5497f1e466da3df3703d4eb3401 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/da979e97b1684b309c707449d157d0bb : SUCCESS in 52m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94c0c50a087b409398882eeb33cfe7ed : SUCCESS in 27m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52f0bc2e1c13436797dd278c9f2ab93d : SUCCESS in 49m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11b7a9d17b7c46f7a39a058c6eb1c1d4 : SUCCESS in 56m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/316b726801dc456b98d0c3ca52b9ade0 : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b133e8868444f509b80ddf9747ffd67 : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c3326790f0740588e47b215aeec2598 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e4aba5f79fe4b69829526ab1b48e468 : SUCCESS in 4m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff119cd21714396b56fd56273b87e2b : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9df59de526644eb9026664edc0157a0 : SUCCESS in 22m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb1f7b55ed1c48e493c35efc4682fb47 : SUCCESS in 35m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18b728b768f34a8e94f0b309b33f15b7 : SUCCESS in 7m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aa0d2fe084fa4cb8827853d52c2eb6c4 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1ac38005d960488882fa02786bacecee : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70ffbbc2bf174e389f3981e94af025b7 : SUCCESS in 4m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/614804002fa04637ba3b2a6775303d7e : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd2fb11966c04782b31a2043a578ad6a : SUCCESS in 3m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67d8842d6e37429b959c636a2ffe5ab0 : SUCCESS in 55m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d9e6d3197744e51916152396c6db683 : SUCCESS in 38m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/220cf5c7228c47dd81953cd07a1a445a : SUCCESS in 38m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84cf8b3f0e854e7dafc4c82c4884331d : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d0a461bfcc4954979e6abe9e2378f4 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/099f5f239ed441b7a5cd67135f7cc71c : SUCCESS in 21m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4246c85b6ae34962b36963a85e5fc7c2 : SUCCESS in 30m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c5f890e8e254bf78fa4ac861909a002 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c7be2c43aed0f58d242c3fa225b5c9206220568b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-02 16:07:18.000000000","message":"Patch Set 5: Code-Review+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b8c36ba081ce821122c89fb3a587f72484ad737e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-03 13:59:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8b9a57666fda932d6a0aab95e67fc34672456d9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-03 16:24:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_36606\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"_revision_number":6},{"id":"3bab0746802398a367b38b32b98a0bd19a68a690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 17:04:58.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e120e26c7b1243f68d96bfef157d90ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c497e321869409792fa748bf22e08c1 : FAILURE in 40m 26s","accounts_in_message":[],"_revision_number":6},{"id":"1d20a1104228c71ad3985bac3de1525f2071d863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 17:31:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acaf017f729b4c34b050ac2f1c2873cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554289362a9e422e83828c7ef5f1253c : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/584f521041204b0b80fe60a69cd18fb4 : SUCCESS in 51m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e00bcd7e24a43419348171bc17c9fea : SUCCESS in 52m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f1be691d0104ffeb15ba35bd15af6ce : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75944800302e48d28aafc180cd0ee3a0 : SUCCESS in 48m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f46c74b86e942deb87ded7088aac46b : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8035eb457574d7dacad1d7888f510f3 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78afbe0cd4874a998962bea594515941 : SUCCESS in 18m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d267b49e0ae4efe97862be70069e6e7 : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef840c5875674e269400bdc81599394e : FAILURE in 10m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9396827eb2ef49888d2f9716efba205d : FAILURE in 21m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ffd6413ba2dd48b5840768f94109de36 : FAILURE in 13m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5466a06a9c404b3192b4d6dcd2c772f6 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1122ec4356f448279d2f19884542d254 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a9f95e1167cd4d3e9c10ae08b3f5ca46 : SUCCESS in 9m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7909e6beb544631ae1f5d21e10ea3fc : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75c8364ea05744979d20bd1fc91c775a : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5df3e849e73a439fb8d8e3518713c5bc : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3693ad63120474d8b3075fd0f6a660f : SUCCESS in 1h 02m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/209e1e6b3b3d425ab5df88db3554c1fc : SUCCESS in 33m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2c84594ee5345e1bcd35e08dd128bb0 : SUCCESS in 41m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccf4259cc87d450980c44082becba9f6 : FAILURE in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6fbf45f6244d32a628c5e26528c5af : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ddc344d0a87494eb6e303049e904895 : SUCCESS in 20m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/681c2f7fa4a2454cad8f87f7fd7dc977 : SUCCESS in 46m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfe2df6be9e24e8a8e209c9f531fa1c2 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":6},{"id":"03977de64c518950607691e9967ab369805712b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-04 13:58:44.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6aaac7a4060055afc133cba13eb8b3f7bc7b04c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 14:53:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e4b80d31b7a4165b911fb8d30116e1e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/862a87c3761e4a8f98f5a913b1342458 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":7},{"id":"e4ef301a4f96b414846e0afbab252599b78cfcb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:09:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b4b312ab0f4c23b781ac1edf97e2f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa6a955d0ac47b2b0f172bcf1b68039 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/acac5324165441ae924d5111c970fa34 : SUCCESS in 51m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9174cd3791b64382a9ebb03d9604268e : SUCCESS in 52m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce647a2cb9744d19b2cbc33271653fef : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62fc6c20e66147388af3cb29e807026b : SUCCESS in 48m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06fb45b1689a4bbb9e4c22d5e4246ffa : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3fcdd4a96e04802a64c62f37eb79b5c : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf79cff931174f75a6c2c211970d0630 : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/445e77527578480a8aea3901b820a014 : SUCCESS in 4m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e61fab5209f4b349ac2e774c1a68f5b : SUCCESS in 20m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4ab7969f25244c07ad9f047cdbed22eb : SUCCESS in 18m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c06b1cee6134497a4e9cb78baea59a0 : SUCCESS in 21m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5131478f738343719e0b7f50e6091929 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5777ce2a5580420ca418f68bd243583d : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20267fae87eb4e32aec983c15804efb2 : SUCCESS in 14m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d05e510fb44143b392f500a618d2858f : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/299f020759e74be9a68436d3fda95197 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf59fb09f4dd4b66a130488ec010f563 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4d636a06cbc4dfd87b73fcdbabd2bb6 : SUCCESS in 55m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19e5a583bdbc481097edc6e390542234 : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f26b507d41f04a67a58e7cf5052ccfcc : SUCCESS in 46m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c288de2fa7744f46997d0fedc4aab124 : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad54c667ac54f76bb7a712ec0b15fca : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b5f6becf662467086ed961382dcd515 : SUCCESS in 20m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d41020e632ff43539b364c36a45534c9 : SUCCESS in 36m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2a25e6aa8304501aadf237fe3cc025d : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":7},{"id":"d4f7d8d1da7dc1931d25c528eb55cc5ed1de258b","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-04 16:19:40.000000000","message":"Patch Set 7:\n\n(12 comments)","accounts_in_message":[],"_revision_number":7},{"id":"79e368989cc4119878a3276f8516a69470b681d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 16:23:12.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1e9e5d55cb69c19f72131f1366c6c98d6ef4f01c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 17:03:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f0aa108063bad197f50e32d6c909e3514b6ce853","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:12:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817269a430704ae79c7c3e13fb36c8d1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a711c7263844b178ca5c63355f851b2 : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":8},{"id":"de569bd025273eb4c82bd0ffcbbbdc9e485d9023","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 17:36:54.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a17b033635b7615663808e6bc8196119a5aa03cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:39:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b772d8a789b6439e9738b154068fef09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671c2cfba8f3466aac54e0e3ddf59210 : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/066c312bbb894ee483be42d1670c9da7 : SUCCESS in 38m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6850ed03a1a340c1be3a2662fc21b1ba : SUCCESS in 1h 04m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a21d5184d0724063a8387ba10db4d279 : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e046d7e2ab4a433c90c32be955f26572 : SUCCESS in 46m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbeed6adf1f045e391bc36c5f14c8bf1 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b3d4935b23c4080ba39fde68acc2365 : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ef2ab93cccf43328ff2604b7104b4bb : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a645f830108c458bb2b8136b47a4e43a : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3f1badcd6854deda0a895bafed5c45e : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea0ccf68168f4e3db2ab55b66885fc9f : SUCCESS in 21m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a5f1476248442218fd9a1bc5cf9d6d1 : SUCCESS in 18m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bcb0bceab67546aba2c8eefd5d460041 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7027b9342f074e6987a82ffa017d9bbb : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26faee32926b4b15a4765c575ce201cc : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0cfdcfc46ab14ce4afa02fb402e8d8bd : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/726829d08fed433298c696e3bc5aee18 : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6416f0704284720a2fffa1e467f0ea8 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48f484d82df44fae8a9740acde1eabc6 : FAILURE in 39m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62d97f72664f4f30a8eac8de84540eea : SUCCESS in 23m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43355b93071c4deea41ee99d39a3a718 : SUCCESS in 24m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/129df67642a745b0ac76b118f33b47b1 : SUCCESS in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e65e79ed8cc4e9b988e249fb886af08 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4993cdc2557e40e09bce66160abf9f38 : SUCCESS in 13m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e106fb49407440ebd0a3c60c456794d : SUCCESS in 46m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0c6a6b926194a4d91c8909fd4557bc2 : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":8},{"id":"b47cb6128794c060ae51e171d2d61f73f8d796d1","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-07 16:23:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"314e605fe29c2cda5a19dceacd4aada9f9f50b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 17:18:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0021c6b5ef4246fb91f4b8c1d4fe6cd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c41c06113347f5b930a533570add13 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/0ffcf86eefe94d8d94f60d33f34d550e : SUCCESS in 27m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4edf7659c2ea4605a4283ee8442b4aa8 : SUCCESS in 28m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef16e4543c684cc9b87a6e1192df6107 : SUCCESS in 30m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fca85395e434447485b971c48a2ade87 : SUCCESS in 32m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fddf52810d9b4be0af7312562c1d95db : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa42cd54a4c046919d39d518e7284f0b : SUCCESS in 22m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0adc08f454b4a328cd0e6cd27721997 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf83127db86447fbaec6d3021d45d0a5 : SUCCESS in 4m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/618f1551bc594f679649660d01be3652 : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7f60472f8e4eb198696df6616037b5 : SUCCESS in 9m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cc9b57a976624a879da202b7cf41e72a : SUCCESS in 29m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18d7c07ed07942a991d70703bfb89872 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d70ff060d8a4be5b35058a975280663 : SUCCESS in 13m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1eaa66ce9ab047d9b8c1d3a0d396d656 : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e982bb41c884877848397615849cd67 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ac26ec56b9d42b7b1fe104dd0f4bb3d : SUCCESS in 13m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3bab2b4837a441c8fb7f7be135ec436 : SUCCESS in 5m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d432c9ab6eb4ae38363bc5431c6bae1 : SUCCESS in 52m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d4d438b046641f7b943c09eb3765069 : SUCCESS in 34m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28bc994644474b899f36e5f6f64c8a66 : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e439912b2a460b97ed4c2f1be9abe6 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf5b844bf034f868a1e97d1b90aff95 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0321b6888e7444ea844f6ae99ce2cec1 : SUCCESS in 20m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba2d5da51c614a67bd9e95c7d3c88ae8 : SUCCESS in 31m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eafc736dcddc49719533d1f4e72a4450 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":8},{"id":"ab497886026d1119da7bc9169c12de78c113760f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 17:29:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b1eefc149b4445a8fadaf98ac8f40f0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b838c584490c4b5cb66b33fa05e032da : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":8},{"id":"1e7e1b43e09053af2c64590139dc9c68561e548e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-08 21:23:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"1562c4da2be64c07863becb1c90107e39a6f1631","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 22:11:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c1de6ca47a49619532ce529800193e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8b59275e8114c808ed53f3a90bbf5f8 : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":9},{"id":"20130c4af4d7bb415ae435e5dc5157a4d18f92bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 22:36:02.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4074b0970ef4309835d381fdb75bdf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7327f1a10c4d4c2684cb0843f4233f30 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/f375266cf1994de8a8ff7b27a6cd216e : SUCCESS in 48m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5ea0b3de2804316947e3d6ebccb894a : SUCCESS in 28m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bec89724938f49f69282e06c86bf0d05 : SUCCESS in 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f22449aabf449a4b32050991b6048e9 : SUCCESS in 32m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99288bac413246fa938578bea3ca9a1f : SUCCESS in 10m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9ff9f642e5244b3a729b6aaab736ae3 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35ec2ef4ae87496cbc9d6b8c2ed7610b : SUCCESS in 10m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be6bf31ced4f4ca7837b4992c080d26c : SUCCESS in 3m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/468a22c0d6a143678103330b333f90a7 : SUCCESS in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/73f12a57df9943a0be4dd182fcca8493 : SUCCESS in 8m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1470ad184345402cb5f46f09520a5628 : SUCCESS in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea27ccc7efad4308ae30d2bddf4c51c4 : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a81b464a5a9144c6840b3c61d8d8c2fb : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c41371b70f949a988c27dc60dfec469 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f52d84999c64f80b59289b609dbfbbc : SUCCESS in 5m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a665c5dc0a9d4c1a839d7fe4fd1622ba : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a92b2c436cb64be199ce6ff98a7568e3 : SUCCESS in 9m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/caf0902130db4292990800b6a36d725a : SUCCESS in 57m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a91fc06dc414550b7f838e835c1930c : SUCCESS in 47m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9855fa8786aa4ccea41782baf7f665c1 : SUCCESS in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46437460db1e44a7bd09a6265c3bf337 : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b42afd218e435994253517b0b48552 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a980607d6d7e436a8d20c86328e9fa54 : SUCCESS in 20m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f01c74da3dfb4248a36b9371431da8ae : SUCCESS in 45m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d895b98ac31648aeb093325c08c839e4 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":9},{"id":"b97bb04002b92f8903f2735bc8c25cdd71f67e13","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 12:30:49.000000000","message":"Patch Set 9: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b4910a85dd95406f2b449ac9e3b61371dfccdaef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 15:43:47.000000000","message":"Patch Set 9: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"96529dae6eee5d0bcda15f52ff387d0838b4091d","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-10 15:36:24.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7ea7622491995a46beba988df7fd58b5dba70c5a","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-10 16:30:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1af6eac6ef35b277e79ede074e8d2de2985f38fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-16 19:59:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e276f7fc9566eeeaafffee8b9feec9298816d1a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 20:39:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/502352a4dfac45608240e6b7d562d788\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b438d16e75884e8ba979bf050258c589 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":10},{"id":"6fb527f58ca832d973453a84e74126632bfb3697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 21:23:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d710a5b9cfde4f03b6c7e0c28737f480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ce17b9921d46248c1d620eaaa6ecfb : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/5f728956a94542a8ba9e400b0c9675c9 : SUCCESS in 44m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9eebff2cb91452d8ff2b67648566a48 : SUCCESS in 29m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa0058e41e43451e96841e804b4bd386 : SUCCESS in 48m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c11e1e25bb8346aab7924514e3d46843 : SUCCESS in 51m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e3f8c8c46ee4797b76793c6e2367e30 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54af3359c829419588b0f26ba5b85b2c : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7270aea331b640f5b4c9a6fd78499841 : SUCCESS in 21m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8de09dbff0fa464eaa82d60c0e169d58 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e1e95c4ef1a4da5b7e8a2e5a524e67d : SUCCESS in 9m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb2ef1e03454d2788d1b86957ba33cd : SUCCESS in 8m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45ee7fd7757f42cc8ebf50ebe4f368ab : SUCCESS in 29m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3f708d2edaef48fdbe5a69227ecd459e : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a0f1f4c2d8744a10b6525044a026959f : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cf15a313a7f743d0ab2c4b85dad287b2 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cad2167173554a6389e5b264f79bc01f : SUCCESS in 10m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5140eb428f2846deb716adaa6d42bea5 : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4ddaabd75614b89959ee130e415af6b : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e6081134b194cfba430ea05700a45f9 : SUCCESS in 52m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20b427c1e0ef49e8a7b18a2b01ca67a0 : SUCCESS in 39m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0ec0c68304349d9afba6d00a406a02a : SUCCESS in 32m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2a3da3f19654631bf465f3f7ff6a829 : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cf8eead81f42cebad9b7a4a899b988 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5324146f653d48bf8fc87a3981e79946 : SUCCESS in 16m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e912d51540643029fb7cdc403902bd7 : SUCCESS in 45m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e0b2ac1460e46aaa9a80bff6f946af3 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":10},{"id":"13976e8d466e2a8ff9af981415e79670b04408d0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 15:38:38.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"714ad90bc8ea1177282400ccca29d99dcc0fd9e5","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-17 19:22:05.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9ca04a4b7f7cb7be3cefd04d4a473c844d15ffef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-02 16:20:00.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"cf37b9e6053683acfc80b939bad5a7006dbfb3f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 17:20:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87e3012ea56744159ffad5059e8b9117\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1854e1c99af4e0a95705d911df8c05f : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":11},{"id":"317c12c3f8f7f4351c0b957191dd16020d398c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 17:23:19.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f93948e3ff4e33ab6cc104ec7db98b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a3d5f14ab349a090c60e6d221a22e4 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/9856f30bea6447b6b6c4ab52d19f558b : SUCCESS in 39m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cedfedbb5e694192978b579da38fe611 : SUCCESS in 58m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6171d8a40bd40b59571afa0b1129b44 : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e782456045f497e88a8006eee66c5c2 : SUCCESS in 55m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d34a114f04e44819bc62f9d8fc7e891e : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2693408996e044f087f88b190dc90535 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db33b0f2ef014c78a207edc76ca9c6b0 : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47393c342f6243d7b6f2333ef199d90d : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/638faaed73534358a3ceb578eee4fa13 : SUCCESS in 19m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b740aa07af84e5ab86164194b2be1b3 : SUCCESS in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/21fa77ef46834d6a8c5f6bf235887e1a : SUCCESS in 28m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e9dea9ce592e474b9176966e9308baaf : SUCCESS in 23m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f1fcbd182794675aad3d748ca29b9a7 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1fae1c409a684639b278eb39548c82a5 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1efbe8dd5b3541df9ed3e21419c62c7c : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd6c27a30fb4468f9284372ecdc5b7f6 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36bfb7db35914240a2e0d65bcd8fc109 : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/845aacce5e504ffb9927fdd739360aba : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9b536041a6e40778b08e7ea2a6c6069 : SUCCESS in 45m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/798fa5330584437c9aab3b273e46af11 : SUCCESS in 34m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/963b234da24f460aa7768cfba7097193 : SUCCESS in 34m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/325f161202754a30adfcfca3b9fee5b0 : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8efe905e7404e629754ee172dbd4546 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4442dd1559064844a8c2d008712c0648 : SUCCESS in 16m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1dee0fb57f74acea3ce5d2654472fa8 : SUCCESS in 43m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51d5ae26ced44f56adcbe4696bf363c0 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e9c9bd4f9e3f33d92169686420444872a091e26f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-19 11:14:54.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1da9e61bacc48a7d44d2bcce58ed53fb5e12fc31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 11:15:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"22dbff7784a46f3d62844834a275ed1c0bc06e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 12:20:00.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd867798c70049ba9cbfd28729624229\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b327445bd1c42c8b91c9b6d27f4bb61 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/09c7b43931454000bc4d65528db1762c : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cda4d8fd49f24d1ca3f471a442a27b65 : SUCCESS in 48m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e870654cefb940558e2bdfd5525c786b : SUCCESS in 1h 00m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76c824a7a48d4817a8e929f30647c3de : SUCCESS in 1h 00m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/792e74886a014a30919a1331b6104fa5 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/278ce18e494544b9a82c6ba0293b942c : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22abd580f70844ab821b209fb1929642 : SUCCESS in 22m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea7bb8f03baf4cc0a3440fab14d556c4 : SUCCESS in 19m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/13b1b337480e4213bb469daecd170cbc : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30434632abcd4d5a96bb8b6a6d31bfc3 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/02da6eabbca647e78c26f218d1c34433 : SUCCESS in 13m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c2647bd879d4adb88dcf51e740b14b4 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e676fd13fa75411d8279e7cf91bd1817 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/266bb3e07679444388db57955cdf5974 : SUCCESS in 50m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f3d513881a14e1588ddde364d55ddd8 : SUCCESS in 44m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e5f1fd023a1407b88bdc9d7f4f540b0 : SUCCESS in 50m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81db1e2402614b75accdfbed21fcf318 : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480569ed8e4b40f4bc3c2ffbbddc62ff : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c1c10712ad943e9ae1cc3d493dd319f : SUCCESS in 16m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274975e1ede147188bf8f4e16bcdc4e9 : SUCCESS in 45m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83d44af3c0d749d0924107eefb0e8158 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":11},{"id":"1cd300fbfa03a25b36b7fbc2a87905f5f52c9394","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 12:20:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"355117ac981dd39cb9a8dd01075109be19b6f19a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 12:22:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79491d5d4ddf4a14a51a0f1c4f891dc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61357bbd0f0472ab8f0f8a266630b04 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/69fcedd9b051485c8030bb8b36e33b12 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"ab6e05922fab390af77c0ed5020fe79b9ab9f6d3","revisions":{"c629695b0269f0a9b0fd1a945936e2b88c3b5df0":{"kind":"REWORK","_number":1,"created":"2025-03-14 19:59:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/1"}}},"commit":{"parents":[{"commit":"2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-14 19:18:54.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c629695b0269f0a9b0fd1a945936e2b88c3b5df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c629695b0269f0a9b0fd1a945936e2b88c3b5df0"}]},"branch":"refs/heads/master"},"a050f79d8d2b5f00eb6e221f9d021e8c8826eae3":{"kind":"REWORK","_number":2,"created":"2025-03-18 13:58:27.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/2"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-18 13:34:12.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a050f79d8d2b5f00eb6e221f9d021e8c8826eae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a050f79d8d2b5f00eb6e221f9d021e8c8826eae3"}]},"branch":"refs/heads/master"},"66a7ecd2916fbfc9ede6108d51cdfe8f60d9135f":{"kind":"REWORK","_number":3,"created":"2025-03-19 13:59:01.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/3"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-19 13:36:42.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66a7ecd2916fbfc9ede6108d51cdfe8f60d9135f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66a7ecd2916fbfc9ede6108d51cdfe8f60d9135f"}]},"branch":"refs/heads/master"},"daee30683132f06d20438a54e0b54e5e8b71523f":{"kind":"REWORK","_number":4,"created":"2025-03-20 13:58:58.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/4"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-20 13:35:39.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daee30683132f06d20438a54e0b54e5e8b71523f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daee30683132f06d20438a54e0b54e5e8b71523f"}]},"branch":"refs/heads/master"},"fd59017ed7b5efddf2b8111f8c57eea5019fd7db":{"kind":"REWORK","_number":5,"created":"2025-03-21 13:59:14.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/5"}}},"commit":{"parents":[{"commit":"38319c6c11b4e210c3816038c333c839b79c286e","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38319c6c11b4e210c3816038c333c839b79c286e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-21 13:54:09.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd59017ed7b5efddf2b8111f8c57eea5019fd7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd59017ed7b5efddf2b8111f8c57eea5019fd7db"}]},"branch":"refs/heads/master"},"bd08a5b0534b6d4430635f574b87d6100e68e3ff":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-04-03 16:24:06.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/944506/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/6"}}},"commit":{"parents":[{"commit":"12fded3227b3392ba614e0681afd4dc8eb02d2c8","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fded3227b3392ba614e0681afd4dc8eb02d2c8"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-03 16:24:06.000000000","tz":0},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd08a5b0534b6d4430635f574b87d6100e68e3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd08a5b0534b6d4430635f574b87d6100e68e3ff"}]},"branch":"refs/heads/master","description":"Rebase"},"0f1ff00d1864faa162e96b92e64bb392ceb2090f":{"kind":"REWORK","_number":7,"created":"2025-04-04 13:58:44.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/7"}}},"commit":{"parents":[{"commit":"832a31fab8e970901b805f1bcfc140b7216a8b0f","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832a31fab8e970901b805f1bcfc140b7216a8b0f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-04 13:33:43.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f1ff00d1864faa162e96b92e64bb392ceb2090f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f1ff00d1864faa162e96b92e64bb392ceb2090f"}]},"branch":"refs/heads/master"},"81cf98c51ce5043c072873f01386039d6dfd72ff":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-04-04 16:23:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/944506/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/8"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 15:59:03.000000000","tz":60},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cf98c51ce5043c072873f01386039d6dfd72ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cf98c51ce5043c072873f01386039d6dfd72ff"}]},"branch":"refs/heads/master"},"5286ab2e0004136fd372d2b720804b8c19f50ff3":{"kind":"REWORK","_number":9,"created":"2025-04-08 21:23:49.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/9"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-08 20:35:12.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5286ab2e0004136fd372d2b720804b8c19f50ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5286ab2e0004136fd372d2b720804b8c19f50ff3"}]},"branch":"refs/heads/master"},"64cd3bbbbb834b5da55f4beaa513d12645be61c7":{"kind":"REWORK","_number":10,"created":"2025-04-16 19:59:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/06/944506/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/10"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-16 19:59:25.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64cd3bbbbb834b5da55f4beaa513d12645be61c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64cd3bbbbb834b5da55f4beaa513d12645be61c7"}]},"branch":"refs/heads/master"},"ab6e05922fab390af77c0ed5020fe79b9ab9f6d3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-05-02 16:20:00.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/06/944506/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/944506/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/944506/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/944506/11"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-12 19:59:35.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-05-02 15:33:43.000000000","tz":-240},"subject":"Object-server: add labeled timing metrics for object REPLICATE request","message":"Object-server: add labeled timing metrics for object REPLICATE request\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I155572bc5f4d955c918c951d6d2af9dbbae0136d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab6e05922fab390af77c0ed5020fe79b9ab9f6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab6e05922fab390af77c0ed5020fe79b9ab9f6d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
