)]}'
{"id":"openstack%2Fswift~944952","triplet_id":"openstack%2Fswift~master~Ic504841714bd607cb9733b2de5126092a79c1094","project":"openstack/swift","branch":"master","topic":"fix-s3api-chunked-transfer-error","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-21 19:36:40.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-21 19:13:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic504841714bd607cb9733b2de5126092a79c1094","subject":"Let WSGI know the length","status":"MERGED","created":"2025-03-18 23:14:11.000000000","updated":"2025-04-21 20:11:43.000000000","submitted":"2025-04-21 20:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"944952-fix-s3api-chunked-transfer-error","meta_rev_id":"55449bb01236da32fe64fd6e6bd92839a6d38c72","_number":944952,"virtual_id_number":944952,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-21 20:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-21 19:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-21 19:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-18 23:48:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-03-18 23:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-19 00:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-19 16:11:20.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-04-21 19:13:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4bf055430e134206f4f9b8241a2f147d470dec6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-18 23:14:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8606d58d7527959ebf2288e32ac42d6d35aba6c4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-18 23:48:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f93a11e239f29831719fa2a6d852407c308f062b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:49:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221de4decd314661a4c91e0327e3d7d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0638c8ea400b458f80e8811fbec845f4 : FAILURE in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d231694c67ae2246eeb9689f12a4f2bf308afac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 00:29:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f99fdb31d094bfcad28290a4404cd3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adeca84d8d14e7c9f1d67a9404e2183 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/066ec47421ae4ebdb6dab07eb3c18898 : SUCCESS in 1h 01m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59209212ecab4631a529e2384b77071f : SUCCESS in 1h 04m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56c67459e981406bb80ef26266883cfc : SUCCESS in 53m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abaf25093c314223bf170246c73f8586 : SUCCESS in 56m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f1b6f62e2094600800a18b3d8835d3e : FAILURE in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8ce517dac924506915e59843612f577 : FAILURE in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/720ccac4be424b9b8cde8c72e8119970 : FAILURE in 10m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/306efa9546a84716a9c1f4a6d89dca26 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a9675a76b364dc49432dc4e447846bf : FAILURE in 18m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06b0713a616d4822b97f5d39376bf15c : FAILURE in 21m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51c5e3d750034e2c81329692e9f511e1 : FAILURE in 19m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/26402bd03a16497291cda7ac643a5f01 : FAILURE in 7m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ab891d7807047f0af75ea0f41030546 : FAILURE in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1906f63d4ce842dbbcd1f600494373e2 : FAILURE in 12m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70dc097c2e7a4e3f982983c30581a417 : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aaf5628141224dc389978038155e9d87 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/11455e24a6b04701bddcff11c43acb21 : SUCCESS in 5m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5157446ef337492a99a39d30b8f62df8 : SUCCESS in 33m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4c440ea449b452baf834b948e066d04 : FAILURE in 21m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66e0d43b97bf4cf49d841171e7ef0e49 : FAILURE in 14m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a3d0bb893bf4307ba557b964e172cbf : FAILURE in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5ee955895f4add80d5a63631dcf212 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0610e28d9114256bb6d6c213f21bbc9 : SUCCESS in 14m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95ea831a55fe49c28bb18143729f45fe : SUCCESS in 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/403b06c0a493422aa3e8dde3667d8189 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c24205e8be2bf8e112440ee510c9f32886894746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-19 00:32:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c695aec17a20f155bc5205a475cf6afc25db4de8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 01:12:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be73f61b731c4bc2af9ab01f19cecd88\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/189d95ed15b147e48190786122fe09eb : FAILURE in 37m 17s","accounts_in_message":[],"_revision_number":2},{"id":"eda4c3bc13e358dc0bdc0c81c84305ef06c37f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 01:35:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cdd3a8da2834daebbe10f4b6f7e75c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5273f5d447f48cd8a2814c1121b682d : SUCCESS in 4m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a75d3e5c7dba4aeab48102f561036346 : SUCCESS in 26m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f367533195a4becb73bbbc751f4ed29 : SUCCESS in 25m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58d3127f4743491d8c59fdb112e4c564 : SUCCESS in 31m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9e2f73b016f41db9300aef316d70d85 : SUCCESS in 1h 02m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a23d78ff3f524f31a0e547a9c8eda8b2 : FAILURE in 12m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a249151e732a4e548d14840dd942eb15 : FAILURE in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e803460e2a434264b4ed65909466bcac : FAILURE in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e049ade2d8e46abb82136d1e4068a80 : SUCCESS in 3m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e744779910fd4243bde738bf1cf3a2cd : FAILURE in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f19424278f6d4d03b266b1ed318bf7d9 : FAILURE in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eb58821fb5594c4c989dc034a42a2578 : FAILURE in 15m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8da8c9de7c6f4e319bcd8916caead2ff : FAILURE in 20m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40256fb814014a1282c63d64663d1b8a : FAILURE in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6fc51fa6343f410abe56c1d18f717881 : FAILURE in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6388f717608d4b7cbc141d097d89e9bc : SUCCESS in 8m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/328e9a8a7bdc4e7f839855ad03f59dda : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06a986ed9f9f4a6bb3e9bf85a21f2c67 : SUCCESS in 9m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7708549b3395447f91acbfc744779d82 : SUCCESS in 38m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f059bb4956fe475087a72cede14b13c2 : FAILURE in 27m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ac56e25a6194bfb86e34565b1a6c1ed : FAILURE in 26m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27291e7610bd4071a706512083091a7c : FAILURE in 24m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56075d4286f4376a624b8631f4e3457 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f7f26795adc4df4870db4d2147bfe22 : SUCCESS in 21m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/825c1fd5abc94a77a4f119a6ed864f5b : SUCCESS in 47m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad1a7a2896943899e1ad77aa2fc860d : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4a27b81ee6c016f3014d0b1dabc21ee160491a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-19 01:59:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"48267f30023916b7abea02e8530636dad8b9bda6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-19 01:59:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cdc0b6cb067167dc5815a23bca123a75da0f9e9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 02:39:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121d1d36186e4b28814049d7fe5089fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e135400463e404295b2bf8365960b7a : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":3},{"id":"36dbf9c80a6df081461a72fd9d68746d14ae831f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 03:13:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d97048dc1b4cfbb61a45eb7fb367d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52cdd3b9e3941baa43aa2a4fabb85d4 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/ebe2b595c6eb4d1fb24df091a57dcb42 : SUCCESS in 1h 08m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad2e4a449d3f48e189d06cf018c25884 : SUCCESS in 58m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61a26149af6744528ecc8ba2618c7e1b : SUCCESS in 59m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c79ac03fa384fae9ec07e02550f4340 : SUCCESS in 1h 09m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91846a56e574408aa812edef90560d3f : SUCCESS in 25m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6694a9bcb314766ab41ef0a5acc5b87 : SUCCESS in 23m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3349956763be41f4b6b89332e6a9353b : SUCCESS in 24m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a818e28e3e0b4965a7fd269b5061ecf6 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/778ee63157124a94a76e4aef2c4380da : SUCCESS in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/113e2ef368144ff1ae66d0da61576ddf : SUCCESS in 18m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a3e9f970ec2443209d8b329a5f49549d : SUCCESS in 18m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/219eaa5caa2c48d0b39ed4edae4f8c1f : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ff5c1adabeb4008b3e2ae0f3c059b04 : SUCCESS in 8m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/28683a61a370435da3ba091639198297 : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcaf12deb0854ec2a2b6ff777144eef1 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55b417e166534535b181506a62ed8462 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0db19a69020d4f29a199b2aa615ac95c : SUCCESS in 4m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/89acada64f934446b3d2bba371c01b48 : SUCCESS in 1h 04m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69c84a7f26634de79c935440dd820bb7 : SUCCESS in 43m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fcf53353699e4e5bac0d77017d999c9b : SUCCESS in 18m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2cffb54eaf64474a2b4be178ca40b5d : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fe9a4daf6d434b8c4e0ee49ad92bab : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d83f26d191ce4fa187179bbe1da79de9 : SUCCESS in 11m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad4d972d945149b2b23c17c4b6947fff : SUCCESS in 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4136d14175e74d74b79ad51d219a57bd : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":3},{"id":"7f60417523ea9cafd499f9f1c907dc2c72484611","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-19 16:11:20.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"110be822106394e12fe49e1a9230b4541098f88e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-18 15:00:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"19e0e1ddeb8f87107825c5cda3298ea34dce93e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 15:33:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdf2f1ed5014248b3d206193862a566\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b538a68898804f488f948dd6a0943dae : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4472f1db86d0f3ec66a162665e367cbe94c62d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 16:08:29.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af3a8585bfa44d949447bb8da0743fb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffeac6205a44db7b4d88fcd9ef12304 : SUCCESS in 5m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/4cb1bba43d0f416a9e23bf368799fd5a : SUCCESS in 1h 00m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8bd98ee9a87429aaf61ea72479e88d2 : SUCCESS in 59m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9edbc28fceb5424387563ee4d9a31946 : SUCCESS in 50m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e983f1659ca34b29a7399f942401732c : SUCCESS in 1h 02m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/108bd7ceaef34e9b9ad74653119c0dd4 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80709e83ddf84bc2b5dea504bbed793f : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0298f4301ef948a8ab0f2010aa0bbb2b : SUCCESS in 21m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08b55a997fdd40e8b8ec00309543edf0 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b2b97eb39194e3287f7a86131768caf : SUCCESS in 10m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d69a8714b64406b59120697a86c964 : SUCCESS in 14m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c566634a82b4817aea948e898fbb525 : SUCCESS in 16m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2c3e0e018784fba9ff2aa670f50d6c6 : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4094078d2437497985a7d7ffe1a77281 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62176f6597934ff59e1bb320ecd4d401 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/804bf9a07aa1439bb93d5d1ede628197 : SUCCESS in 5m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e7eacae03454e32a162079e228d9e93 : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a862a67a870e4b0980130d7ffcd71ac8 : SUCCESS in 3m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e0f7063cab44949a9ebd0d7b8b2c001 : SUCCESS in 48m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef4510c854cb4b7eb75ed308f7239672 : SUCCESS in 18m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b62f4ecd5764f59ab1566bbbd7014a6 : SUCCESS in 36m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66c7a6acd0654931930ad666c189901a : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead5a226d6424120a4a368fd4ab8ec8f : FAILURE in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17a87271a0a64ab6a662631eeade6792 : SUCCESS in 8m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d713ebc8812844c6a08d8ccbee09a2b9 : SUCCESS in 48m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ed7243ecb00445aa0f71018d881c4f2 : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5c32b70712f90f307970e580c688701ebe5656fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-18 16:27:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f4baae8587d8988e74b52ec723d64a5101da2235","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 17:13:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1039fa7a85204ba4807f9203365b6b21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27bef60802df4cf69db12a6a4562038e : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":5},{"id":"a7ce8e5d06ea847449a0f678948ac80ae1d928ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 17:19:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d9e0e977e4d4ccc98c8233af6b2062b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a2f380fe824f28aa8bbe4eb5b52101 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8d81365516f24f78982e9b030787494a : SUCCESS in 39m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92ae72c055f34d66bc656e20370e01e5 : SUCCESS in 28m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2757a2e580848ad858b48fb2767336f : SUCCESS in 34m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d0fde3e3ae64cb5987f29d01d2d44af : SUCCESS in 32m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7c903e541b04f9284ad57c06b83a00a : SUCCESS in 9m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2640c42b6576407188105845d7fb907a : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d73b3ff4ab064ef3804cc9029df66e62 : SUCCESS in 15m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae08eab76494451db6f6b7e7a31a211f : SUCCESS in 11m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd51278383db45399508b99e88e1fad5 : SUCCESS in 20m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46b2961ef18647f2995f48b7b38ee6b4 : SUCCESS in 13m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/084b5a799ed34bcba2b34d74acbe85cc : SUCCESS in 23m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92e364b132d246c3874c998b45b39749 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a59f039dd272439dadd1e78e512209cd : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6a27390ca9b48e59c12614aac961b5a : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47e6e53e11c541faa1cdfd4021a6b5ea : SUCCESS in 5m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1ca8d5068b148d987bf5c67a1789aa4 : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e2dc6d78ab8451bbbd2ab96323b4b7a : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb181fe6d62344a58ddc5418f2e26788 : SUCCESS in 45m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/962a15dfa3fd4f5ba7b8cf2f77d215d9 : SUCCESS in 48m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e284d06b3344d68ada1c187e4d3d904 : SUCCESS in 31m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2077c75c329496fb3954ac429b0aa8e : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a38ca88b306940679b0ab182ddda2e63 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5b4654148d5473daed14624c47dad56 : SUCCESS in 16m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6065146fa4d94cc797c5c4d65aa672d1 : SUCCESS in 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95d50ba7b71d4983a6ee2c03fc1c7662 : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":5},{"id":"9f59f550551cdfe7d9d9de93e77abb178ad4c53e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-21 19:13:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"142197aec0eb0119ae86e6b4e2448705e347f4bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 19:14:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4bf84584cb693407552238637122d8bf85647dcc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-21 19:36:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42abe6936a0038db623d2cb415dc89cbdbda0ce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 20:10:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df99dc3891b34e09804d2f384f9562d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc00a1c975147a3bf24d5b859aadbe5 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/61c99a7679094600b8c852475899f395 : SUCCESS in 28m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bb41ca773cd439d82fabfca573931cb : SUCCESS in 27m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcec3b15f0c14cab83cbffc8f8e739e8 : SUCCESS in 30m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/774037ddfdbd4dd8bede94cdad817248 : SUCCESS in 31m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/918430214355414c8076b97c8f7d7a84 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/258c63f87d4c4e109d0737009f61f627 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4bf4907ee054a25af5b45d9656217a5 : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9222548bd3a8499ba341c646e470cbf0 : SUCCESS in 31m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/97fbfa4b5a8f4db58b97abf7871d7eb1 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62131479e3304da98bf8539cdabc98d1 : SUCCESS in 12m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe93d3ddea2347eb97bcb30e9041edef : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f376cb42445c4630a724ec17a0ec62de : SUCCESS in 7m 02s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a0346b103d44022b7d3624f556ebcc3 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aacd05f303f44b3791c7e4565dd51830 : SUCCESS in 55m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bc8cd53ddc04f0db8d412c51f03ab68 : SUCCESS in 35m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/570ed15c1ccf440b9de3c3fc0878cfc3 : SUCCESS in 33m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f3b557b83b64cb88ad2598d36fc222c : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13fdac6e76244029fd0c25a2cedcd72 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87d75602314d4c5e9c040b1adf2d8291 : SUCCESS in 22m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73dcdf96557a48749cd8898cbd2e1194 : SUCCESS in 39m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4de4da08b70647239f1770ef0eac41cf : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":5},{"id":"08001ecbb08fc288cd9b3fc70eba396fab250041","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 20:10:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"55449bb01236da32fe64fd6e6bd92839a6d38c72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 20:11:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc7065e5a97e48de8d2077a10661a848\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ae45126357460ea3f392f0de46f65f : SUCCESS in 1m 09s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7b2b3363ee0f4173bb7d9f1c8e665e36 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06","revisions":{"83af430dc624c37910c2288edbd8712db6e198a4":{"kind":"REWORK","_number":1,"created":"2025-03-18 23:14:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/944952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/944952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/944952/1"}}},"commit":{"parents":[{"commit":"aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:12:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:13:59.000000000","tz":-300},"subject":"make s3api errors not chunked-transfer","message":"make s3api errors not chunked-transfer\n\nChange-Id: Ic504841714bd607cb9733b2de5126092a79c1094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83af430dc624c37910c2288edbd8712db6e198a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83af430dc624c37910c2288edbd8712db6e198a4"}]},"branch":"refs/heads/master"},"0d0fdfbec0b0c35a5a883e7bcdb1f49cca51e5b4":{"kind":"REWORK","_number":2,"created":"2025-03-19 00:32:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/944952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/944952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/944952/2"}}},"commit":{"parents":[{"commit":"aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:12:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-19 00:28:30.000000000","tz":-300},"subject":"Let WSGI know the length","message":"Let WSGI know the length\n\neventlet.wsgi likes to check for hasattr(\u0027__len__\u0027) when setting\nconnection close on expect-100 errors, but uses \u0027content-length\u0027 in\nheaders when deciding on chunked-transfer.\n\nWhen we know the length we can support either interface.  Also we can\nimporove s3api to return error responses with the content-length known.\n\nChange-Id: Ic504841714bd607cb9733b2de5126092a79c1094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d0fdfbec0b0c35a5a883e7bcdb1f49cca51e5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d0fdfbec0b0c35a5a883e7bcdb1f49cca51e5b4"}]},"branch":"refs/heads/master"},"4e5ad467b2fc19f32fc641bda24891e5e93533ca":{"kind":"REWORK","_number":3,"created":"2025-03-19 01:59:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/944952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/944952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/944952/3"}}},"commit":{"parents":[{"commit":"aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aed3257c4ba4ccac55f9d7bb5bf9f428257f66a2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:12:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-19 01:58:10.000000000","tz":-300},"subject":"Let WSGI know the length","message":"Let WSGI know the length\n\neventlet.wsgi likes to check for hasattr(\u0027__len__\u0027) when setting\nconnection close on expect-100 errors, but uses \u0027content-length\u0027 in\nheaders when deciding on chunked-transfer.\n\nWhen we know the length we can support either interface.  Also we can\nimporove s3api to return error responses with the content-length known.\n\nChange-Id: Ic504841714bd607cb9733b2de5126092a79c1094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5ad467b2fc19f32fc641bda24891e5e93533ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5ad467b2fc19f32fc641bda24891e5e93533ca"}]},"branch":"refs/heads/master"},"e8e79aa5d8b1959863fa29c96b041086b0b03288":{"kind":"REWORK","_number":4,"created":"2025-04-18 15:00:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/944952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/944952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/944952/4"}}},"commit":{"parents":[{"commit":"fd9ceecc506f88cd940edd7bb369574c044c5038","subject":"Merge \"s3request: refactor to introduce SigChecker classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd9ceecc506f88cd940edd7bb369574c044c5038"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:12:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-18 14:59:18.000000000","tz":-300},"subject":"Let WSGI know the length","message":"Let WSGI know the length\n\n... so that eventlet.wsgi will include Connection: close headers on\nExpect: 100-continue error responses from s3api and make aws clients\nless unhappy.\n\neventlet.wsgi likes to check for hasattr(\u0027__len__\u0027, resp) when setting\nconnection close on expect-100 errors, but uses \u0027content-length\u0027 in\nheaders when deciding on chunked-transfer.\n\nWhen we know the length we can support either interface.  Also we can\nimporove s3api to return error responses with the content-length known.\n\nChange-Id: Ic504841714bd607cb9733b2de5126092a79c1094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8e79aa5d8b1959863fa29c96b041086b0b03288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8e79aa5d8b1959863fa29c96b041086b0b03288"}]},"branch":"refs/heads/master"},"1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06":{"kind":"REWORK","_number":5,"created":"2025-04-18 16:27:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/944952/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/944952/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/944952/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/944952/5"}}},"commit":{"parents":[{"commit":"fd9ceecc506f88cd940edd7bb369574c044c5038","subject":"Merge \"s3request: refactor to introduce SigChecker classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd9ceecc506f88cd940edd7bb369574c044c5038"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-18 23:12:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-18 16:27:35.000000000","tz":-300},"subject":"Let WSGI know the length","message":"Let WSGI know the length\n\n... so that eventlet.wsgi will include Connection: close headers on\nExpect: 100-continue error responses from s3api and make aws clients\nless unhappy.\n\neventlet.wsgi likes to check for hasattr(\u0027__len__\u0027, resp) when setting\nconnection close on expect-100 errors, but uses \u0027content-length\u0027 in\nheaders when deciding on chunked-transfer.\n\nWhen we know the length we can support either interface.  Also we can\nimporove s3api to return error responses with the content-length known.\n\nChange-Id: Ic504841714bd607cb9733b2de5126092a79c1094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
