)]}'
{"id":"openstack%2Fswift~945116","triplet_id":"openstack%2Fswift~master~I20ab80a93038e0ad60fb6437e28b2e739098d509","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_bytes","hashtags":[],"change_id":"I20ab80a93038e0ad60fb6437e28b2e739098d509","subject":"sq? fixups for proxy_logging","status":"ABANDONED","created":"2025-03-20 15:17:09.000000000","updated":"2025-03-25 11:25:31.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9d1a08aa7f716d7cebcd72c1536345c0c97a2bde","_number":945116,"virtual_id_number":945116,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-20 19:39:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 16:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-20 16:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3670fbbc3a39cdc9dad2898fe9fc8fc55df87205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 15:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81c4651ffc12720410b5860fb34de10ed674627f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 15:34:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cfe2a8ae8ea37232872128ab51cf98b3dbf813c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 16:26:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df6fdf6ca7b840c1bc6ad9c3ae83723c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cc8db6a881e498a9d9deacfdc653721 : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":2},{"id":"54fb9dacf2a39e696969d8f89f75f142454e003d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 16:40:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50641a5de9b14c7fb6aba6414f6f7379\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06300ef49a34f8d9666a38178911142 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/6276852994314a96a664b15e75856999 : SUCCESS in 52m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/281db81ab389458b8f2734589d05ec6b : SUCCESS in 51m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/555f8da568004c4ab3472102163779f8 : SUCCESS in 45m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69f67c002cfd45dda00e7371724cd184 : SUCCESS in 58m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f3496b9912a446a84bbb71a48d8765e : SUCCESS in 25m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5809bdcc1a1c4954b035a4b0eb939049 : SUCCESS in 25m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/690968eef4794aa1b33b2fcded996efa : SUCCESS in 28m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c968dfc1c5e3470eac8281ec7011ea45 : SUCCESS in 3m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82e65d59fe204170b2bde0222df3689c : SUCCESS in 9m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/734ac2172dc047eaa767edaf33969b95 : SUCCESS in 23m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8121d41cf7534f2ca92fe8482ba54d93 : SUCCESS in 33m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/13585189b4ea438288929dd6c760e547 : SUCCESS in 7m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44b2751e62d64854a08410d7267acc51 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f818970ba03e41239b0c6e970b781ff7 : SUCCESS in 21m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c092856777dd487e9e6f2cafc30411a1 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2535da816124268825cb4e1ffb27fc1 : SUCCESS in 7m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/60dcf74ed2f5446998218677be46cafa : SUCCESS in 5m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ff10e11e4e8430faea6f28ca40931d9 : SUCCESS in 56m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/712a00f298544dc1a98778febaf3a30d : SUCCESS in 44m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fb78883fa3a49b1bdc2273f012895a7 : SUCCESS in 19m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49d664cb985e498b89ef2d2b6cfed2d1 : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9463937107554f97bd4d3bb4a404d8ca : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfb4c3971e674fa4a97abfb0ee19599b : SUCCESS in 20m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/763adcc4471c4206a6a6efd3471ca5d4 : SUCCESS in 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da7f1416123c425dafdf20eb9f667312 : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4f4f59146db38d1a0a489fc5b5a654086cc9700c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 17:03:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88c49d1a637681590311e48abd7d621c8782b39a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 18:27:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e137d66d9c5fc66fccac816cc4ca18782c31bd64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:12:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf08fa022da4639969484da7d888da1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b53daa288d364fcb882b0514bb262217 : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":3},{"id":"603d72d434235d993fa31b06b5126e9b40f1ada1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:39:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01c85d396e64f40a0055d2a3a1004ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4946eb41afb145b8ae013cf7d8127489 : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/c1ab92c43c3845e09c86413aa0ce0650 : SUCCESS in 30m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a90ccc4ed5947bb8877b2e312922222 : SUCCESS in 30m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43a4233291c94cb994613b698c9dc7b3 : SUCCESS in 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/301201190cb546ae85c86f2aa2ee22b4 : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f3dedea6304937acca4ab09bce819b : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/989660eaa78142d8a331e75b4d0bb6ad : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27fb284cba944d49a2476b9a6f0be012 : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecfb666ec0bc4734af4b31caf68fcab2 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9853d4d9eb7e4f00bbf7399d2584dd84 : SUCCESS in 17m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfb0e04f2acc4dc3b1e83cf85e0b53be : SUCCESS in 34m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a29df313110343ee966b8aabea5536ea : SUCCESS in 22m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e0850bfc122441191dfcf4245281fee : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a0cccd9dfd743ee94c8dc06ecf05456 : SUCCESS in 25m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af32233ea0864902b88413438e349656 : SUCCESS in 13m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f8e5376530f4eca9db3be15476d611d : SUCCESS in 7m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ad2bad183d041a5bf13e3381ef35b56 : SUCCESS in 24m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46ab8022c5a8431ba362239ce3d34366 : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad4f11ba9d39443f9eaa8f012171c9ab : SUCCESS in 1h 08m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6cb60bd0ba6c4c0996108d2842f413af : SUCCESS in 39m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d2b152804ba45f5ba6e7bd6f88f346f : SUCCESS in 34m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca5ae249fc784359ab1da0e8287bfa9c : SUCCESS in 30m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f577a7c69d14f42b372d9b8c1d024d2 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd420769661d4c46ba81c194a6c6c55a : SUCCESS in 10m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7884303e5661459689f5d91b27b9a272 : SUCCESS in 37m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a57559e406b4ab18cb49387c8f3b002 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9d1a08aa7f716d7cebcd72c1536345c0c97a2bde","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 11:25:31.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b85864f9b1c54867d6095628554cf339c76ba6c","revisions":{"0cad7b55fa7a3b49f429436cced31e825326244f":{"kind":"REWORK","_number":1,"created":"2025-03-20 15:17:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/945116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/945116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/945116/1"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 13:53:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 13:57:37.000000000","tz":0},"subject":"sq? fixups for proxy_logging","message":"sq? fixups for proxy_logging\n\nChange-Id: I20ab80a93038e0ad60fb6437e28b2e739098d509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cad7b55fa7a3b49f429436cced31e825326244f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cad7b55fa7a3b49f429436cced31e825326244f"}]},"branch":"refs/heads/master"},"71343bb6e2520706caa6231e69ef04c85d4e5c69":{"kind":"REWORK","_number":2,"created":"2025-03-20 15:34:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/945116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/945116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/945116/2"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 13:53:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 15:34:14.000000000","tz":0},"subject":"sq? fixups for proxy_logging","message":"sq? fixups for proxy_logging\n\nChange-Id: I20ab80a93038e0ad60fb6437e28b2e739098d509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71343bb6e2520706caa6231e69ef04c85d4e5c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71343bb6e2520706caa6231e69ef04c85d4e5c69"}]},"branch":"refs/heads/master"},"6b85864f9b1c54867d6095628554cf339c76ba6c":{"kind":"REWORK","_number":3,"created":"2025-03-20 18:27:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/945116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/945116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/945116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/945116/3"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 13:53:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-20 18:24:58.000000000","tz":0},"subject":"sq? fixups for proxy_logging","message":"sq? fixups for proxy_logging\n\nChange-Id: I20ab80a93038e0ad60fb6437e28b2e739098d509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b85864f9b1c54867d6095628554cf339c76ba6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b85864f9b1c54867d6095628554cf339c76ba6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
