)]}'
{"id":"openstack%2Fswift~945228","triplet_id":"openstack%2Fswift~master~I3b48303cf725ab1100193f096f625ce2ff1f9312","project":"openstack/swift","branch":"master","topic":"p-statsd-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-25 10:36:27.000000000","reason":"Change was abandoned"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-03-25 10:36:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3b48303cf725ab1100193f096f625ce2ff1f9312","subject":"sq? fixups for labeled statsd api","status":"ABANDONED","created":"2025-03-21 15:18:40.000000000","updated":"2025-03-25 10:36:27.000000000","total_comment_count":9,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"758604e891e3f57bad73683fd12c3839cb393517","_number":945228,"virtual_id_number":945228,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-24 18:16:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-21 16:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-21 16:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 19:22:25.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"}],"messages":[{"id":"400a433f0ad5dc3f95e892294b56bc38e2c5ff4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-21 15:18:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8fe228e4a53b24b74918140ee93d450920b4ffa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:08:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635c870ae8734e2c9d6a75f1cdbb895a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45b96a811d1044ff883bcee5972b4218 : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2936bfb93a76f6a703770a3fe7655d7c8bfd7d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b5f4743524465ba64b8d9be3fcef28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a440e003ca92440bbc387c063a7d5973 : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/86f2c3fd81a243638ae21ddafcceb16e : SUCCESS in 54m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88db3840bc1d4f3990dca5926ae47f5c : SUCCESS in 51m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01cccaa0193c44298835cb81dfccd879 : SUCCESS in 47m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158d3c3d3eb346ff87c3fa5c5c2f7051 : SUCCESS in 32m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83e8fdfa84b5409db0da8356b9262ffa : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a055e70fdc44b30b5531e71016b4f4a : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c86efb3d1ac1411e87477b4b0cad8151 : SUCCESS in 12m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1eb719036b7847a9be1b9a7b05aa4724 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84873e6eea384854b8ba71c4a28997aa : SUCCESS in 7m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3343f747ded5460aa584624138966d77 : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2b9fc38bd0e4482bb9b315ada3f1755a : SUCCESS in 28m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a55ae3c8e2d54083a46bda9d8115aad6 : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d3fb9ab2b1843e89878b3bbb9d387aa : SUCCESS in 12m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/645e68a361164b29af4a9f5aba66f446 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5cbb365aeba4fdbbe8bf5ffedcca1be : SUCCESS in 4m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0046442706f94c82a76f7c0ea025f10a : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b2bbf11a05f4196908f9e1fac2ea432 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68725d1acc2040a9827a357845b72ef2 : SUCCESS in 52m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91f76acbd71e48c19cfcdf706f3843a9 : SUCCESS in 37m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/775e30895f4b48a5b8be54f91cf0ec37 : SUCCESS in 18m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08721ee3572a4877861b537a02bfd9bb : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5082c2f8208f4152980332a16c555240 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd36be18218647658a09dace4b4a36fc : SUCCESS in 11m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee3d28fd9df6456ab260bae80dd2eab0 : SUCCESS in 38m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01f778aa348e42a296c724baa15fab64 : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fc7228ea126377263d5bd13f73efa0b41b5fd55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-21 18:43:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0a82e34d55f32ad3f27e90f691e3c1c8bdb4520","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-21 19:22:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cebc8f206f501162d617bd0e965f272ad6288eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:29:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e62e0eac994b628464659e2a67ba28\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c140f0a8aae74d5498415548878f5c54 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4618f7147a5e25415c8322cb15cd0e3083b626aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:44:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/434dc76f5d7441ccb704b5fbe3f89313\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9de799a49af469994d0b1df130e353a : SUCCESS in 5m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/56632fe785764ee5b9a4111337472599 : SUCCESS in 32m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5dbee9df98054e5daf26df116cb8576c : SUCCESS in 46m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/556a9e5f07bc4df59f07fc3b244a3acf : SUCCESS in 47m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/366b99f51bee4af097a77b06e3fb1b3c : SUCCESS in 51m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41082b9edee54eacb98aee1e5d565491 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d23d63d1f6743b5a734085b4615fc94 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b41572c571194b1d98493b5e0efc2269 : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d422aeeac2b4884ad44ae7b3073987b : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/839899e5b06949b886a878025b98e68f : SUCCESS in 18m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b7eed8957c924eb4ada1202daa31edf7 : SUCCESS in 13m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/df254d88c20443b79f384194e16a2652 : SUCCESS in 26m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0283b8b7f7c44880a2d22d3359dcdf6f : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/00075185fb7b4d098353e6ada60abd69 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5c0c14603d1409aaf9b3fad8e9a586e : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6439ef9b37f4296a465e83fdeedbe35 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fe4ecfca6614ae9bf82a8c4b6220d70 : SUCCESS in 13m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/16de81f59015457fa2bb53143ceeae93 : SUCCESS in 10m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/15d2f9c69c0c4cbcb0448b8df81c62fd : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8817292f2c442c4a202dcdf92a1fe1a : SUCCESS in 37m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f11b3c12d3141a29796069fd522c427 : SUCCESS in 17m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4606bd7b79cd4c7a8511d2d561f9dbd8 : SUCCESS in 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ae730079964285ac3cad39d2a3e3f1 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bd7d00fd3f94e5286dfe81289f68c1f : SUCCESS in 21m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/331f2074154e46f5882ed5ce27999dbf : SUCCESS in 37m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c4a66920e3945b6acca7ef7f3e6e7e5 : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5df50c018b1151ae8a45cb2febda42157f3cee6e","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-21 21:18:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4baa6c357625641b0b778a603d639c4677e0318","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-24 13:59:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"35faea5cd0ee0725940b9c7d9b80c67e8b8b48f0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 15:34:14.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4c148d7326855f415b91baa5afc05b9097f92474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 15:37:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ee3ed167808c27f28a4cc1399481dc0c39af81b6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 15:46:43.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4a11c9fd2ba5529bd856a41d365ddaf592c441c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:16:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f5c45b75204f7d9f687ba2f2b2a3da\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0657ceea804546a7bd24a75c7a929b8a : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f0dd26bb2a35ed010e71f6b73873f8915c8494ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:57:03.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03a48d56631e47aeb9fba30ab56604f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1c26b01c6548ff91310c0ff5316497 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/55d1088405c74d669a92fb574b91c36a : FAILURE in 1h 05m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8636e66aba5440f0a6bbc9116a688639 : SUCCESS in 52m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/acf7dc8ea856476fa12fa9192fae9310 : SUCCESS in 47m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/050d0a938b734c5290b4f6d57a2016f0 : SUCCESS in 59m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e777008ae54f4aee9ce0efed728281cd : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de62c706446b45c7adfe34b224fd7732 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83c936699e1441ce8d1fce1bf57498b1 : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3887fef63ef94ea59429ba5ed304e4d9 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56a5cef664144af5b7d4d1afdf927647 : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6dcd7e3fa78f4bf4b17f6f8a8d1ec7ff : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b86131d459fe4e7ea40cd6a8a6d71450 : SUCCESS in 11m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ed614606e334462bf1018f7ae521906 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6379bdca9ff4f7e8f70ed359b2367aa : SUCCESS in 8m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/13f96dcaded54915b25242a062765aca : SUCCESS in 8m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1166bc1abe9b4a4b82a7dccc741c499b : SUCCESS in 3m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6265876d4a94e01b5c0c33296c39323 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ff6c8ac75f64edb910a8612fe585b9a : SUCCESS in 5m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed79263807ef4d788b2ba0b8f5dd9291 : SUCCESS in 43m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aca8585f64d44ab8a94647fb31ae4598 : SUCCESS in 29m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa69e399bbd2439983a093007fe4532c : SUCCESS in 29m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87bbd9c650af4482b1edc4ba81ead954 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdc82bddb95464da89a7a3eadbf78f0 : FAILURE in 2m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5095febc0a946d2ae4b88e6a9cf8c57 : SUCCESS in 16m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ec208296b4049d4b252253121e65ec9 : SUCCESS in 48m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/757887fa70c84cf397465ec21bc088a1 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6e0af81a8d11c4083a6c17186ea18d93ca2e0135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 17:15:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"33d3beffe064dec6fc1e7b44b80a035b87258610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 17:58:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca606d2cb3b441697dd944a9a3b9f56\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/164a5fa47c1f4ddd9631404bd5018cce : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":4},{"id":"b068e25dea3d014c94a60518ac00f6fe26f87f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:16:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37df286906354714975f736f041e978e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09050cdfd946453584ce3e2af0ea9d77 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/87d6b18752b644e985dafdc14d7cd557 : SUCCESS in 57m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c9e919e1cc6446c8087830e1cdb34d3 : SUCCESS in 39m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/625831deaca0463e824de827ff8abd09 : SUCCESS in 49m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/709238a317b44049a16a4cbf4c2f37dd : SUCCESS in 48m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c62ccf42a3c746b2b0994e08afe7b06a : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/998a2978192a4202ab8f738b0b2325a2 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d75f9ae2da64c02a6e6966c0cf9a3e9 : SUCCESS in 17m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66efaa14b86b4282a69b9a8cc9cec54d : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86cab630e8054ed586e645f5cdc79845 : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb9675a396bf48a6b475d5be97591b8d : SUCCESS in 22m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e4d1a9b1098c41c687f4715d7dca38c4 : SUCCESS in 10m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d6c7c7c7459c4c8d9e42aa9a0983ffaa : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce94aca979594206bdf1c7b7af753615 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/182597ce68b34de1a02ad6e22576d5ec : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e97cd0913afe4b0cb3fb5df06fbb3b5f : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae209abc8a2c4f91bb3de082071d83bc : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dbc71029d19b4784a146cb8f0be728c1 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39220b500c6245cd876517c46bb91068 : SUCCESS in 32m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee5577985ea345e99805ad3bdb3432a3 : SUCCESS in 22m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/299b89a8f22c40d4b6007f07e90b25ba : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b32d6827c44c4186a8674053263e382d : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf37bd401464535b24618f7e3823811 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fb980849ef14439bf932ab54d0e45a4 : SUCCESS in 17m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9316c7957ff4aa69b3d14c37bf47fc6 : SUCCESS in 54m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d32860b1c53048de8ff6d09fc9d587e7 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":4},{"id":"758604e891e3f57bad73683fd12c3839cb393517","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 10:36:27.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ef9fc8b6cd0eace55163f3f4ee050b78b1e96348","revisions":{"529a6d8a96639ba97755a587e0116cdd6f0f127f":{"kind":"REWORK","_number":1,"created":"2025-03-21 15:18:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/945228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/945228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/945228/1"}}},"commit":{"parents":[{"commit":"b9ef50c32b28d5b361bca13e599f11e2c8c1559f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9ef50c32b28d5b361bca13e599f11e2c8c1559f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:06:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:06:59.000000000","tz":0},"subject":"sq? fixups for labeled statsd api","message":"sq? fixups for labeled statsd api\n\nChange-Id: I3b48303cf725ab1100193f096f625ce2ff1f9312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529a6d8a96639ba97755a587e0116cdd6f0f127f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529a6d8a96639ba97755a587e0116cdd6f0f127f"}]},"branch":"refs/heads/master"},"137707ce501da74b08d1bd2e385ab84e6369e776":{"kind":"REWORK","_number":2,"created":"2025-03-21 18:43:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/945228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/945228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/945228/2"}}},"commit":{"parents":[{"commit":"b9ef50c32b28d5b361bca13e599f11e2c8c1559f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9ef50c32b28d5b361bca13e599f11e2c8c1559f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:06:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 18:42:39.000000000","tz":0},"subject":"sq? fixups for labeled statsd api","message":"sq? fixups for labeled statsd api\n\nChange-Id: I3b48303cf725ab1100193f096f625ce2ff1f9312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137707ce501da74b08d1bd2e385ab84e6369e776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137707ce501da74b08d1bd2e385ab84e6369e776"}]},"branch":"refs/heads/master"},"9a261efd8d93dd36bb39a9a57d0a56e31e8ca78d":{"kind":"REWORK","_number":3,"created":"2025-03-24 15:37:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/945228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/945228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/945228/3"}}},"commit":{"parents":[{"commit":"b9ef50c32b28d5b361bca13e599f11e2c8c1559f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9ef50c32b28d5b361bca13e599f11e2c8c1559f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:06:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:34:47.000000000","tz":0},"subject":"sq? fixups for labeled statsd api","message":"sq? fixups for labeled statsd api\n\nChange-Id: I3b48303cf725ab1100193f096f625ce2ff1f9312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a261efd8d93dd36bb39a9a57d0a56e31e8ca78d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a261efd8d93dd36bb39a9a57d0a56e31e8ca78d"}]},"branch":"refs/heads/master"},"ef9fc8b6cd0eace55163f3f4ee050b78b1e96348":{"kind":"REWORK","_number":4,"created":"2025-03-24 17:15:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/945228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/945228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/945228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/945228/4"}}},"commit":{"parents":[{"commit":"b9ef50c32b28d5b361bca13e599f11e2c8c1559f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9ef50c32b28d5b361bca13e599f11e2c8c1559f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:06:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 17:05:19.000000000","tz":0},"subject":"sq? fixups for labeled statsd api","message":"sq? fixups for labeled statsd api\n\nChange-Id: I3b48303cf725ab1100193f096f625ce2ff1f9312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef9fc8b6cd0eace55163f3f4ee050b78b1e96348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef9fc8b6cd0eace55163f3f4ee050b78b1e96348"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
