)]}'
{"id":"openstack%2Fswift~945333","triplet_id":"openstack%2Fswift~master~Ic79432946b91e698c7392a28e0a2898709ae3ae3","project":"openstack/swift","branch":"master","topic":"p-statsd-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-25 10:39:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic79432946b91e698c7392a28e0a2898709ae3ae3","subject":"sq? fix labeled metric format with sample rate","status":"ABANDONED","created":"2025-03-24 15:37:29.000000000","updated":"2025-03-25 10:39:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bce61dbec1fe077a3adeaf6f333bbf922e94b19e","_number":945333,"virtual_id_number":945333,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-24 18:25:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 17:49:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 16:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 16:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 17:49:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1976d2e9bf0127427db57398037b3d7f393eef19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 15:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e209f5201955729d7fe702588f1eecda7a6b7f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:14:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9b58cc36964999b9b5fc471a6e0c35\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43a9f80d820548d4906c12231fed9870 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"97942cbe7052449d8c090d6a64fe7eff83ecea43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:42:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d32b4ea131ee41afb263e04fca38843e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd2e7afa28347679be8f32093c1aba7 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f990c04e52fb41a7854b9f4995144f19 : SUCCESS in 29m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a89ecfe8b12447fa4269f5ccba47683 : SUCCESS in 59m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aad49bd67d764da095618db1180eb968 : SUCCESS in 47m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a54094d5ee64232931faf262d386fa8 : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19ff5aced85947fdbb6403a66396d9f0 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f0cf9ae0cc84235a1441b73a1a61cf9 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35c9887792a4403c8e4b470a0f6a4f0a : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e4a88dc883c4741809ca5fc9ea23c75 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16ca63a78c284fb5ae049b6a1463519b : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bb656ed3493435a8f2547f39aff51e5 : SUCCESS in 21m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3409e042dca438aaff808020cb5ae00 : SUCCESS in 23m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2ea77d3c6c0f46e1bbf64cf83b3d5fe0 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cbb3134423ba4805b591460c4f723193 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/da4760651fd24b888a123ff2a81dfae0 : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5668a32d42af4af3b554aea75962d55f : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7569ce3ef0b40c88c17afb778cdc1e2 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3fdecca4c5574a888b7432df64be800f : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3bbe3ae4e884e95b72eb70d6a2e0228 : SUCCESS in 55m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e616bf5e5a294e2e80d400a1e19fc965 : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/687b9be50fb64cdea072eb234159c3b1 : SUCCESS in 35m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ea1ab81af054704a47c3ab12c108bda : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ca2ceec2bf4c0982c3ca3e0c2bb807 : FAILURE in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b890b2ec8574ada8d3fd84550ad4039 : SUCCESS in 16m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af3757825eb94c738b4b3e377ce8a212 : SUCCESS in 31m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1385603f697d4abe91131037a24ca8c3 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2fbc049d922874a56d2482f7cd341c1834b44ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 17:15:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d2f8fab918b9b6d0c4d2f0cee390ef1abf0fe772","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-24 17:49:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e21bb24936c243c7cb7867553e742bc704d0949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:23:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b84f0384764532a3067bd4775a53a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80e2ebb42e1a4b60b813d9e8a1ec32e3 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8776a0e2fc3cb864dc26a613af023212041a7091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:25:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cee42c1b286c448d990922face7beba8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39df4c9ea454bd1ba5c6929820dc84d : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/44e94a71eb9c4bb5b1ff924e46e42bf1 : SUCCESS in 41m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/12a66ab81174495da183df5a30a26e29 : SUCCESS in 44m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/866c594c3fa449108f0ca34193a4a5cb : SUCCESS in 48m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57be06ec08fd49aabf2785e336308846 : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87eac13a75554fd5bc7cc6ed20bbcd20 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a265428db3154424adaf79601279404b : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ade6976c982d472a8fb045a554d1a34d : SUCCESS in 12m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21c941c1df5d43bea5c8f457e32b90f1 : SUCCESS in 3m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab1b62c3cb9c47b783a6bbe58e06b1c3 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17940e36743a4d39b5f26d02fbf3cabe : SUCCESS in 22m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bf53ab5c04a143b580ebf102c01f3f91 : SUCCESS in 28m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2c851d70778a4ee48c1e0433a625e051 : SUCCESS in 9m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/52b869317a18408686e05b2a882924ef : SUCCESS in 7m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a977dc3edcc14fd18621bd76c6016f8c : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ec4b4f242e64a2494750a118159aa91 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e5a9a4df0204dd6845d7be17dd80a42 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/611c6b6d686a4d24bc249154c439f31e : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be46dc2e93f24eb5a025c698380ed714 : SUCCESS in 54m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4265b0dc81a4f7abe5bfc0fc2836b1f : SUCCESS in 35m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b4216da3b9c497fa6a4097af9825a8e : SUCCESS in 39m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec88c02428fa40abab2ef055490c41bc : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8ead50b32b4211ab1d69caa5c6ce5a : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f4a851907c64dad84ba8672208b87da : SUCCESS in 16m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28e6116f5d4c4708b78277ee6b7b64f1 : SUCCESS in 49m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a117060e7904997be3032d5993a7822 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":2},{"id":"bce61dbec1fe077a3adeaf6f333bbf922e94b19e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 10:39:21.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/909882","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6","revisions":{"1495fab5ef134232fe8a9bef1334673f98f56c93":{"kind":"REWORK","_number":1,"created":"2025-03-24 15:37:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/945333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/945333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/945333/1"}}},"commit":{"parents":[{"commit":"9a261efd8d93dd36bb39a9a57d0a56e31e8ca78d","subject":"sq? fixups for labeled statsd api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a261efd8d93dd36bb39a9a57d0a56e31e8ca78d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:09:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:36:16.000000000","tz":0},"subject":"sq? fix labeled metric format with sample rate","message":"sq? fix labeled metric format with sample rate\n\nChange-Id: Ic79432946b91e698c7392a28e0a2898709ae3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1495fab5ef134232fe8a9bef1334673f98f56c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1495fab5ef134232fe8a9bef1334673f98f56c93"}]},"branch":"refs/heads/master"},"1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-24 17:15:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/945333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/945333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/945333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/945333/2"}}},"commit":{"parents":[{"commit":"ef9fc8b6cd0eace55163f3f4ee050b78b1e96348","subject":"sq? fixups for labeled statsd api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef9fc8b6cd0eace55163f3f4ee050b78b1e96348"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:09:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 17:05:27.000000000","tz":0},"subject":"sq? fix labeled metric format with sample rate","message":"sq? fix labeled metric format with sample rate\n\nChange-Id: Ic79432946b91e698c7392a28e0a2898709ae3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
