)]}'
{"id":"openstack%2Fswift~945334","triplet_id":"openstack%2Fswift~master~I9c8aa938d32affacd0b1c1e41237e18edf618199","project":"openstack/swift","branch":"master","topic":"p-statsd-tests","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-25 10:39:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9c8aa938d32affacd0b1c1e41237e18edf618199","subject":"sq? omit labels delimiters when no labels","status":"ABANDONED","created":"2025-03-24 15:37:29.000000000","updated":"2025-03-25 10:39:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f69e8ec1e860a7d7abe7d0e1dc01ec44f98e6dc7","_number":945334,"virtual_id_number":945334,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-24 18:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 17:35:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 16:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 16:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 17:35:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e1f8021fcb9a6e80d364e6861880ec6623e13170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 15:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f277a29cf5a564caf916e885746cde24eaa47c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:37:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e6d0fc60c7480082ff51744fa4751e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c21134f055914845964c8467316efb0b : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b82e634335f366117b19f8f39235119324cd7569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:46:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf012dd349cf4f768993148153e6eda4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356c9b12f1a14e55aae8ba90e864163f : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/85400d928c654ba2b5c8d935380533f8 : SUCCESS in 44m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de4f812db81f4687aa78fd26f58b29ed : SUCCESS in 49m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/950a6feaf5284fbdb614d6ddd4285302 : SUCCESS in 42m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fbaf8e4824f43909283beed45129dd3 : SUCCESS in 50m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/536e6b6eee734955b981713bbf0ad9c3 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6aaddc4bb7ec4a78be287d142454dd45 : SUCCESS in 22m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9738ca3d39734033b1d1f3538b2c885e : SUCCESS in 20m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3e391c658fa4b2b9e9f42e512644e44 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8aa5a37bfd409291c148f294ed3fd5 : SUCCESS in 15m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df8cfaa18d01496abcf086883b55e8d4 : SUCCESS in 9m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/253e874a61b04c35ba4ed1c5bcc1dc6f : SUCCESS in 18m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/99c7c69d8fb3432aa8121d1f8845f42b : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ed28a4b8e2344febb59c917e1af8f042 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d770789d4394142b9535f6a6668078b : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a531af644012410b87e5746b73b4dd42 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e20066509ae94871968c1540549e0622 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d65eadb62854fc38a1c39d4d9af43ef : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/294ae86a13f74d88be92b0d48780950f : SUCCESS in 58m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d340eb2819b418d93f2a51bf22e021c : SUCCESS in 35m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0bc540b96ae40c889079e75fca0d5c6 : SUCCESS in 22m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54568b9e54464cf19b8356cf4ef7d06a : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241fb6adf6b34a9f9a6e0f3c114b80c4 : FAILURE in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dd7c8fbacc0498fa5884cdda7e6dcc7 : SUCCESS in 22m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9d6c7f3f6a848cbafd4ec44a073a716 : SUCCESS in 50m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56cd54209e2348bcb37d8770115a7f40 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"303d6b65bc5b0ae606b8b837a2e830429bd2c84d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-24 17:15:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fd4acd6303c9d93900ff4ccdf790a5443d7383c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-24 17:35:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e13ee047ea846b5c2c8920e2bf317a649b952145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:00:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ed4a524d8b42f190ddeae4f2e770fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3679df526f2542e89fe102cd41769ab3 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ce80cbc1b92d8e9bd304a7db8b09aa0285b871e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:21:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30883e0d45ca4e73b5b86efb9d02f849\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdb33ec998f47a786f86183e6344b20 : SUCCESS in 4m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/beeeefe9d0044956bc852fbb73e8ce36 : SUCCESS in 27m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/274ad5d1a5874111a24336941e09e22d : SUCCESS in 33m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee9041319bfe47cebfb88acf0c64a20a : SUCCESS in 55m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f73394f231d4ada99b7ebbf0e3b4580 : SUCCESS in 45m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccab7350e2e343ee92e985ac1ab8e1b6 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83976d2ad2d74cc5a61a997f8b49bf15 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83d84aff3fd4d918fcfd9fbf35ed589 : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e69460c151549818314843333d8878a : SUCCESS in 4m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d4ce569f3b645fd9b368bfb709b5e5f : SUCCESS in 9m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/321516fe109f4ba4a0deac26d1448116 : SUCCESS in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44b9e6340adc4959865dfd0b088aac2f : SUCCESS in 20m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c15309bed9d34727bb34efaa3d5c7b73 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84bcabe7db594b5da3ab4472432bebec : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2a3aedc5b7c486898e7c95fa4a8dfd8 : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59d5492047384e4998e640ac514ecef9 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bca7935088e0495e8261553b36e6da33 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dbccb9c7d2664df1a0d359d76676d486 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/282ef82deb7d48d98aaa95314af7fb5b : SUCCESS in 47m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b849c96880a46b6be688699a8cb6301 : SUCCESS in 36m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a977340c5b648599d8bb330f6ff98c2 : SUCCESS in 47m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b185d16f17347e6b3e573c82dfbf7a4 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d083b8353646f784d63070ff6c320f : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36c1b626601048f8835427709f827443 : SUCCESS in 8m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b871a44af594b66b2aa849792f6b244 : SUCCESS in 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a3f14c309043d6b562b4ed32fadbad : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f69e8ec1e860a7d7abe7d0e1dc01ec44f98e6dc7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 10:39:15.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/909882","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da0f9852674d6a0d696aa7a9cb9150b420f29883","revisions":{"18b38df56a74a0fea4e38405e3c8627ae9153537":{"kind":"REWORK","_number":1,"created":"2025-03-24 15:37:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/945334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/945334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/945334/1"}}},"commit":{"parents":[{"commit":"1495fab5ef134232fe8a9bef1334673f98f56c93","subject":"sq? fix labeled metric format with sample rate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1495fab5ef134232fe8a9bef1334673f98f56c93"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:10:57.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:36:16.000000000","tz":0},"subject":"sq? omit labels delimiters when no labels","message":"sq? omit labels delimiters when no labels\n\nIf there are no labels then revert to the simple metric form.\n\nChange-Id: I9c8aa938d32affacd0b1c1e41237e18edf618199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b38df56a74a0fea4e38405e3c8627ae9153537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b38df56a74a0fea4e38405e3c8627ae9153537"}]},"branch":"refs/heads/master"},"da0f9852674d6a0d696aa7a9cb9150b420f29883":{"kind":"REWORK","_number":2,"created":"2025-03-24 17:15:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/945334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/945334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/945334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/945334/2"}}},"commit":{"parents":[{"commit":"1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6","subject":"sq? fix labeled metric format with sample rate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f7f9e1bd3d7087b71dc90f6380e07bdacb1c3e6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 15:10:57.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-24 17:13:10.000000000","tz":0},"subject":"sq? omit labels delimiters when no labels","message":"sq? omit labels delimiters when no labels\n\nIf there are no labels then revert to the simple metric form.\n\nChange-Id: I9c8aa938d32affacd0b1c1e41237e18edf618199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da0f9852674d6a0d696aa7a9cb9150b420f29883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da0f9852674d6a0d696aa7a9cb9150b420f29883"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
