)]}'
{"id":"openstack%2Fswift~945648","triplet_id":"openstack%2Fswift~master~I219bbcd88667cdb85274a26e52bd1217237c8744","project":"openstack/swift","branch":"master","topic":"no_dev_id_1","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-26 20:01:03.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I219bbcd88667cdb85274a26e52bd1217237c8744","subject":"DRY out array resize/normalize","status":"NEW","created":"2025-03-26 18:55:32.000000000","updated":"2025-03-31 20:21:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"787224429ee9f25ff1c30e5bae567d3b125a92fb","_number":945648,"virtual_id_number":945648,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-31 20:21:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 19:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 20:01:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c18b10e2844ba19024eee27b13e41786b76c0f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-26 18:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9a301ef74555b5531c059be99582d0fe1d9620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:47:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a724eb1c02304259bf049fb92e0cfa73\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c600e039b8c40cd807475178a4328e7 : FAILURE in 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"20e70c446a4fa5959b65194a9a0cc4da0aa6ca79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 20:01:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3a1c3fe3f184447bfc8b392ff8f5799\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480ebd236a1543f6a0ec8b645b54c407 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/764cf7132dea43ffb02a132106c11250 : SUCCESS in 58m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52d41456abd749bd913cbb1033ce1578 : SUCCESS in 53m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a74c82718ea248c98b6a6d1553fbe61c : SUCCESS in 42m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8090d2dc0cde4ef092f58ff9759a8f13 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5856f2a21f7e4a058dc14cb4dd349caf : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cba6521d8e74195af8656cd2e7ae569 : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98e8b5fa0ceb481dbfd111f46ac19df7 : SUCCESS in 12m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04d4f4ff33d54256a9f2cba0f2396fd8 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e60b534a6d5e47b7bc6127744c9091b5 : FAILURE in 10m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/07d549c044284bb991a0521d80b25342 : FAILURE in 16m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8f3e8b9e49d462abdf0d3e46d02d3fd : FAILURE in 26m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c5f8d05ab950440487ec72d38cefae83 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b460843caba46408ebfb571a94ce4a2 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/260a01576cd740acba6082b908e5ab0d : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/497c214a15a84a0aabac28d8cb3906ac : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bdebd5e13974a3a9dd517f173f45e8c : SUCCESS in 10m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e77fa31d4564af995f8f65b606bc4fe : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ffd88760913408e9ac50ef032d7816e : SUCCESS in 38m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a17d90d724324d998fe8f6e707e78871 : SUCCESS in 37m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c08062057ee24422861a02b986299da3 : SUCCESS in 48m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cd87572aeb94b58a45e2a2dc3fb1bbb : FAILURE in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44fe058ee604ce185a4419b2bc7c1d6 : FAILURE in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/059a14dce68b4ead85b9440cb5e26c20 : SUCCESS in 20m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da78aae0714d488e93ab6a2c74b81bd4 : SUCCESS in 58m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e65daa84930b4e539c3724db1beb11d1 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7f2b014a1d4ceddb3813fabefa6f7fa477be7e58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-03-31 19:20:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8124cf91367330175c3a2da5e592e14fc18fd1d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 19:54:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d67a43149df4c3fb1a01da1e8996d3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12cd50417aa744ae9d285e89d7016723 : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":2},{"id":"787224429ee9f25ff1c30e5bae567d3b125a92fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 20:21:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f3075e754ce47d0af78849fe3fad7d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5963017efb42239c3c62ff218a3e4b : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/2f546d13d4ab4298b8eb992e218fe3da : SUCCESS in 49m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c23e55625f414dd3ad1b5dd1e6019ae3 : SUCCESS in 51m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c36c6ba3ce6546bda383092d6906e8ed : SUCCESS in 48m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c7f8310cc44ddea2f8eac5b543ef64 : SUCCESS in 58m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3874db0fee134509997c588a89308e73 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8262a97ae36948408877429dc7f707e0 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a8dc47491f2458a8b9c72c903ac55a7 : SUCCESS in 16m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ac7a0a60b3b408d8c3c8e716fdb91e6 : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4525b2e4c084425ca9edc76e3366a67a : SUCCESS in 12m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14ae6a27084d4a848e0147697fe1282c : SUCCESS in 16m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ecb18bd1e5441e38aa6c0f40508bea5 : SUCCESS in 18m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/63cb43a8a2254d73a88b6f7df85a6332 : SUCCESS in 7m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a5341e6203244c39a27a9b76eaf06109 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c0843b1549e4606a2fac2c2fa80d01c : SUCCESS in 11m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a55ec589fe0f427b9b03ca4448b7d305 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aecbcabdf1b74565939fc4d7c4033c5f : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/41334f6255544f6890453c06cf142c11 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40900f2aef2b4438823d1694320e8dd3 : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9049d8e436dd4c648198c8a271026e98 : SUCCESS in 37m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/950075c684704d809056addf6d68452f : SUCCESS in 37m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f680fb9cfd6045729f710d903805ed55 : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10577504409444d69a5b673772ffc9e4 : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60d745fc550144ea8040a4e3e87b2b53 : SUCCESS in 19m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/778aff6ab18d476fb327298a491bf5b0 : SUCCESS in 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cb0ca4f382b466a9745047f84e5a27d : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d07c6d74aaeaede515008c6ed4f8805cf79b70ff","revisions":{"91a40d3b904a7936e4f3dfacbf80c43b62421ed1":{"kind":"REWORK","_number":1,"created":"2025-03-26 18:55:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/48/945648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/945648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/945648/1"}}},"commit":{"parents":[{"commit":"1a46e96ec814b024169cdf4a38fc299359186f15","subject":"swift-ring-builder: add --dev-id-bytes to write_ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a46e96ec814b024169cdf4a38fc299359186f15"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-26 14:07:38.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-26 18:53:22.000000000","tz":-300},"subject":"DRY out array resize/normalize","message":"DRY out array resize/normalize\n\n... and remove unused bits:\n\n * RingValidationWarning\n * RingData.set_dev_id_bytes\n\nThis will apply some consistency and sand down rough edges.\n\nChange-Id: I219bbcd88667cdb85274a26e52bd1217237c8744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91a40d3b904a7936e4f3dfacbf80c43b62421ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91a40d3b904a7936e4f3dfacbf80c43b62421ed1"}]},"branch":"refs/heads/master"},"d07c6d74aaeaede515008c6ed4f8805cf79b70ff":{"kind":"REWORK","_number":2,"created":"2025-03-31 19:20:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/48/945648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/945648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/945648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/945648/2"}}},"commit":{"parents":[{"commit":"1a46e96ec814b024169cdf4a38fc299359186f15","subject":"swift-ring-builder: add --dev-id-bytes to write_ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a46e96ec814b024169cdf4a38fc299359186f15"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-26 14:07:38.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-03-31 19:20:20.000000000","tz":-300},"subject":"DRY out array resize/normalize","message":"DRY out array resize/normalize\n\n... and remove unused bits:\n\n * RingValidationWarning\n * RingData.set_dev_id_bytes\n\nThis will apply some consistency and sand down rough edges.\n\nChange-Id: I219bbcd88667cdb85274a26e52bd1217237c8744\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d07c6d74aaeaede515008c6ed4f8805cf79b70ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d07c6d74aaeaede515008c6ed4f8805cf79b70ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
